)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~660266","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I0adc25c01ce2b3d6233cb5e92e90d094627356bd","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"opendev","hashtags":[],"change_id":"I0adc25c01ce2b3d6233cb5e92e90d094627356bd","subject":"Install python-httplib2 on lxc-hosts from epel repo","status":"ABANDONED","created":"2019-05-21 07:21:51.000000000","updated":"2019-05-23 06:23:25.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0961fe4d393370a51754d28ce810036b803e7569","_number":660266,"virtual_id_number":660266,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"date":"2019-05-22 12:20:37.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2019-05-21 08:06:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":-1,"date":"2019-05-22 12:09:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-05-21 07:34:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 07:34:53.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-21 08:06:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-22 12:09:21.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"}],"messages":[{"id":"9a14a7a37fe6104b726e712280843b6427be959b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-21 07:21:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c2282166b384e333c6f524e04b4c70d3fd6ce2b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 07:34:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b41acc12369a738e91b2fd4d1ff66020e56cf79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 08:06:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/66/660266/1/check/openstack-ansible-linters/c1a5f48/ : SUCCESS in 6m 57s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/66/660266/1/check/openstack-ansible-functional-debian-stable/5e4f6fe/ : SUCCESS in 7m 22s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/66/660266/1/check/openstack-ansible-functional-centos-7/f1ad54d/ : SUCCESS in 10m 22s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/66/660266/1/check/openstack-ansible-functional-ubuntu-bionic/71360fe/ : SUCCESS in 7m 08s\n- openstack-tox-docs http://logs.openstack.org/66/660266/1/check/openstack-tox-docs/03eba1b/html/ : SUCCESS in 2m 31s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-dir-centos-7/ae2a462/ : SUCCESS in 14m 34s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-dir-opensuse-150/7f3492a/ : FAILURE in 14m 09s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/6dad112/ : SUCCESS in 11m 13s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/8b10bd2/ : SUCCESS in 12m 44s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-btrfs-opensuse-150/664b1b8/ : FAILURE in 20m 02s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/f8d8b40/ : SUCCESS in 10m 42s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-machinectl-centos-7/d01d212/ : SUCCESS in 13m 36s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-machinectl-opensuse-150/5dc4201/ : FAILURE in 14m 25s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/0714dfe/ : SUCCESS in 12m 31s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/df2d0f5/ : FAILURE in 15m 11s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/66/660266/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/265a574/ : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8c1f7f2f4b942a43d2fc4c9ef0ed3b63712b8fb1","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-22 12:09:21.000000000","message":"Patch Set 1: Code-Review-1\n\nAgreed with evrardjp on some points.\n\n- We want to install this from extras, not EPEL\n- We can do enable extras with include pkgs in the tempest role instead of doing it here\n- It is not outside OSA, the tempest role requires this package, but we can enable this repo with includepkgs in the task right above where it is installed in os_tempest.","accounts_in_message":[],"_revision_number":1},{"id":"590945c6efa031d91df759d51ddb479dfd947448","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-22 12:20:37.000000000","message":"Patch Set 1:\n\n\u003e Agreed with evrardjp on some points.\n \u003e \n \u003e - We want to install this from extras, not EPEL\n \u003e - We can do enable extras with include pkgs in the tempest role\n \u003e instead of doing it here\n \u003e - It is not outside OSA, the tempest role requires this package,\n \u003e but we can enable this repo with includepkgs in the task right\n \u003e above where it is installed in os_tempest.\n\nBased on the failure http://logs.openstack.org/42/654942/10/check/openstack-ansible-functional-tempestconf-centos-7/4b8943c/job-output.txt.gz#_2019-05-21_08_27_10_139112 and http://codesearch.openstack.org/?q\u003dpython-httplib2\u0026i\u003dnope\u0026files\u003d.yml\u0026repos\u003d It is required by os_nova and os_neutron for distro devel packages, I am not sure os_tempest is the correct place.\nWe need to enable extra repos.","accounts_in_message":[],"_revision_number":1},{"id":"a1906bb56dc1c50cf112c6725aec596b10b27cfe","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-22 12:20:39.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/#/c/660697/","accounts_in_message":[],"_revision_number":1},{"id":"7a0009d4a2bb2dc0e25c0ebb8abd9700a3536fef","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-23 06:23:25.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eff57f75ea907cf6f0209ec4a664d2037b66a304","revisions":{"eff57f75ea907cf6f0209ec4a664d2037b66a304":{"kind":"REWORK","_number":1,"created":"2019-05-21 07:21:51.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/66/660266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/66/660266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/66/660266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/66/660266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/66/660266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/66/660266/1"}}},"commit":{"parents":[{"commit":"a6cae27fa3e6d03b48ba34468df4af90c77f4880","subject":"Merge \"Revert \"Temporary fix of failing aria2 on centos\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6cae27fa3e6d03b48ba34468df4af90c77f4880"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-21 07:20:18.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-21 07:21:48.000000000","tz":330},"subject":"Install python-httplib2 on lxc-hosts from epel repo","message":"Install python-httplib2 on lxc-hosts from epel repo\n\nCurrently on CentOS 7 jobs, python-httplib2 package is not found\nthe enabled repo on lxc_hosts and it is available in epel repo,\nlet\u0027s install it and make the job passing.\n\nChange-Id: I0adc25c01ce2b3d6233cb5e92e90d094627356bd\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eff57f75ea907cf6f0209ec4a664d2037b66a304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eff57f75ea907cf6f0209ec4a664d2037b66a304"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
