)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~664296","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fpike~I53a9bb1ccd520b0fc6bf3ae085a2dfc9010b33dc","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/pike","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I53a9bb1ccd520b0fc6bf3ae085a2dfc9010b33dc","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:43:36.000000000","updated":"2019-06-10 17:21:03.000000000","submitted":"2019-06-10 17:21:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664296-1560187263257-d9d96933","meta_rev_id":"4cfc0c9d66024b4e104cff01c5485336511019e8","_number":664296,"virtual_id_number":664296,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 17:21:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:02:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:02:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 17:21:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f78b720897f0947bdac4c9580af41681ff4512d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:43:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e672644394ecf3a9dbbc783c3d034fb1ef66232","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:01:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/96/664296/1/check/build-openstack-sphinx-docs/e6d11f1/html/ : SUCCESS in 2m 56s\n- openstack-ansible-linters http://logs.openstack.org/96/664296/1/check/openstack-ansible-linters/fe0f2ff/ : SUCCESS in 5m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/96/664296/1/check/openstack-ansible-functional-centos-7/e2ae906/ : SUCCESS in 11m 37s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/96/664296/1/check/openstack-ansible-functional-opensuse-423/70a1e6b/ : SUCCESS in 7m 41s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/96/664296/1/check/openstack-ansible-functional-ubuntu-xenial/d3acce2/ : SUCCESS in 7m 20s\n- openstack-ansible-lxc-dir http://logs.openstack.org/96/664296/1/check/openstack-ansible-lxc-dir/8d1f50b/ : SUCCESS in 13m 08s (non-voting)\n- openstack-ansible-lxc-btrfs http://logs.openstack.org/96/664296/1/check/openstack-ansible-lxc-btrfs/f060a4c/ : SUCCESS in 11m 35s (non-voting)\n- openstack-ansible-lxc-machinectl http://logs.openstack.org/96/664296/1/check/openstack-ansible-lxc-machinectl/1108a14/ : SUCCESS in 11m 24s (non-voting)\n- openstack-ansible-lxc-overlayfs http://logs.openstack.org/96/664296/1/check/openstack-ansible-lxc-overlayfs/d8f03fa/ : SUCCESS in 12m 11s (non-voting)\n- openstack-ansible-lxc-zfs http://logs.openstack.org/96/664296/1/check/openstack-ansible-lxc-zfs/1a921c9/ : SUCCESS in 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cc0e491588ceee3c08658da1e332ac5b6423d74","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:02:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51afdb39f58a1078c27a5cb43f31fbda5d7384dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:02:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9632b1ccd6f804784e1ea0b253113f662b0dffad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:21:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cfea21287b1d6867053effec6fbd6af4e3b30f1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:21:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/96/664296/1/gate/build-openstack-sphinx-docs/7418062/html/ : SUCCESS in 3m 40s\n- openstack-ansible-linters http://logs.openstack.org/96/664296/1/gate/openstack-ansible-linters/dd9dfc7/ : SUCCESS in 5m 48s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/96/664296/1/gate/openstack-ansible-functional-centos-7/fe14b38/ : SUCCESS in 12m 30s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/96/664296/1/gate/openstack-ansible-functional-opensuse-423/d47f69c/ : SUCCESS in 7m 59s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/96/664296/1/gate/openstack-ansible-functional-ubuntu-xenial/59e9c60/ : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12793c336ba7680fe35f93c56552942761fb50d9","revisions":{"12793c336ba7680fe35f93c56552942761fb50d9":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:43:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/664296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/96/664296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/96/664296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/96/664296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/96/664296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/96/664296/1"}}},"commit":{"parents":[{"commit":"4c73718fac7d8ef54d9d32bbc18a1b0b90401c0e","subject":"Merge \"Use pkill for lxc-dnsmasq systemd unit file\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4c73718fac7d8ef54d9d32bbc18a1b0b90401c0e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:35.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I53a9bb1ccd520b0fc6bf3ae085a2dfc9010b33dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/12793c336ba7680fe35f93c56552942761fb50d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/12793c336ba7680fe35f93c56552942761fb50d9"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
