)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~664574","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fqueens~I8da5de0074b5ffe470a585bb3ae5394a1dae9efa","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/queens","topic":"lxc3","hashtags":[],"change_id":"I8da5de0074b5ffe470a585bb3ae5394a1dae9efa","subject":"Dynamically use lxc 2/3 syntax","status":"MERGED","created":"2019-06-11 12:53:24.000000000","updated":"2019-06-12 21:59:26.000000000","submitted":"2019-06-12 21:59:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664574-1560376766405-b1f8f8f4","meta_rev_id":"40cccd07b2e4d986bbe3c95d5446c6ed81ee9787","_number":664574,"virtual_id_number":664574,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-12 21:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2019-06-12 17:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 21:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 21:20:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 17:29:57.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-12 21:20:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-12 21:59:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"938312794a707e562265eb8f10e2e1664d91845c","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-06-11 12:53:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9582a33f2f61f26aa58603175e1d7039f750069","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 13:14:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/664574/1/check/build-openstack-sphinx-docs/babde12/html/ : SUCCESS in 3m 09s\n- openstack-ansible-linters http://logs.openstack.org/74/664574/1/check/openstack-ansible-linters/6cef904/ : SUCCESS in 4m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/664574/1/check/openstack-ansible-functional-centos-7/6abf4d3/ : FAILURE in 7m 24s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/74/664574/1/check/openstack-ansible-functional-opensuse-423/06adef8/ : FAILURE in 5m 44s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/74/664574/1/check/openstack-ansible-functional-ubuntu-xenial/d2b882e/ : FAILURE in 6m 03s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/74/664574/1/check/openstack-ansible-lxc-dir-centos-7/a03af11/ : FAILURE in 14m 02s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/74/664574/1/check/openstack-ansible-lxc-dir-ubuntu-xenial/022b8e2/ : FAILURE in 6m 51s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/74/664574/1/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/6858b62/ : FAILURE in 7m 07s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/74/664574/1/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/2050a40/ : FAILURE in 6m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/74/664574/1/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/db39709/ : FAILURE in 7m 35s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/74/664574/1/check/openstack-ansible-lxc-zfs-ubuntu-xenial/dfc2b81/ : FAILURE in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bec07dd7008bb015744e82d979e8e227dab87427","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-06-11 15:28:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6720140a758a716a8eb00ab3e26ae999b014561","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:57:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/664574/2/check/build-openstack-sphinx-docs/b010307/html/ : SUCCESS in 3m 50s\n- openstack-ansible-linters http://logs.openstack.org/74/664574/2/check/openstack-ansible-linters/9b64b21/ : SUCCESS in 5m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/664574/2/check/openstack-ansible-functional-centos-7/373fbcc/ : FAILURE in 6m 30s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/74/664574/2/check/openstack-ansible-functional-opensuse-423/0b78911/ : FAILURE in 5m 24s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/74/664574/2/check/openstack-ansible-functional-ubuntu-xenial/293bea5/ : FAILURE in 5m 10s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/74/664574/2/check/openstack-ansible-lxc-dir-centos-7/b01859a/ : FAILURE in 9m 45s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/74/664574/2/check/openstack-ansible-lxc-dir-ubuntu-xenial/8e88c35/ : FAILURE in 5m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/74/664574/2/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/2cec9ad/ : FAILURE in 6m 39s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/74/664574/2/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/addf1e7/ : FAILURE in 5m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/74/664574/2/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/a065df5/ : FAILURE in 6m 35s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/74/664574/2/check/openstack-ansible-lxc-zfs-ubuntu-xenial/3dfb6a5/ : FAILURE in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"fa1f26f78c54cf70d04bcc2363bb81b4f717b390","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-06-11 17:23:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c2005780b852dda525459ffa36f2c37004b94561","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-06-11 17:49:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f134b68f71446ec5aafb1426a4e0a5eda2da482f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 18:28:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/664574/4/check/build-openstack-sphinx-docs/942aa20/html/ : SUCCESS in 3m 02s\n- openstack-ansible-linters http://logs.openstack.org/74/664574/4/check/openstack-ansible-linters/1cada12/ : SUCCESS in 5m 55s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/664574/4/check/openstack-ansible-functional-centos-7/9626882/ : SUCCESS in 8m 52s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/74/664574/4/check/openstack-ansible-functional-opensuse-423/310c514/ : SUCCESS in 7m 30s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/74/664574/4/check/openstack-ansible-functional-ubuntu-xenial/585e063/ : SUCCESS in 7m 57s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/74/664574/4/check/openstack-ansible-lxc-dir-centos-7/b16025c/ : SUCCESS in 19m 27s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/74/664574/4/check/openstack-ansible-lxc-dir-ubuntu-xenial/5736650/ : SUCCESS in 11m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/74/664574/4/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/550a7fa/ : SUCCESS in 10m 32s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/74/664574/4/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/785fd87/ : SUCCESS in 19m 13s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/74/664574/4/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/a3f0db7/ : SUCCESS in 12m 28s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/74/664574/4/check/openstack-ansible-lxc-zfs-ubuntu-xenial/1f2ee77/ : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":4},{"id":"56f686081a0da82f7b770fde05eb493445b570ea","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-11 20:44:13.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"69500c9c63469a924c7228abcce6f9c901b35bc1","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-11 20:45:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7349bb7482aebc5fac4eceae0c083247a2ed96e7","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-06-11 21:18:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3f445a055f3b3938d1ccdad5de6331dd083a91ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:00:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/664574/5/check/build-openstack-sphinx-docs/5c379bb/html/ : SUCCESS in 3m 34s\n- openstack-ansible-linters http://logs.openstack.org/74/664574/5/check/openstack-ansible-linters/df8894c/ : SUCCESS in 5m 36s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/664574/5/check/openstack-ansible-functional-centos-7/002d6f8/ : SUCCESS in 9m 16s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/74/664574/5/check/openstack-ansible-functional-ubuntu-xenial/351b400/ : SUCCESS in 6m 42s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/74/664574/5/check/openstack-ansible-lxc-dir-centos-7/3f5080c/ : SUCCESS in 22m 15s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/74/664574/5/check/openstack-ansible-lxc-dir-ubuntu-xenial/e5a024e/ : SUCCESS in 13m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/74/664574/5/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/7055852/ : SUCCESS in 13m 32s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/74/664574/5/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/1575b3e/ : SUCCESS in 16m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/74/664574/5/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/2c01c85/ : SUCCESS in 13m 56s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/74/664574/5/check/openstack-ansible-lxc-zfs-ubuntu-xenial/8e9a18a/ : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":5},{"id":"b41bf7d2d7daea8dd783e7c491a7f39415f6f129","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-06-11 22:13:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e73346c3f57cc561430f70c01d73e7e25d450e85","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-06-11 22:29:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"76f1287f71f44dc509f86d92dcf14c86282b7555","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-06-11 22:30:03.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4321a038651dc4655a68f3e2f5dc1edb8bf069fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:53:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/664574/7/check/build-openstack-sphinx-docs/7cebbd0/html/ : SUCCESS in 4m 55s\n- openstack-ansible-linters http://logs.openstack.org/74/664574/7/check/openstack-ansible-linters/9b97045/ : SUCCESS in 5m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/664574/7/check/openstack-ansible-functional-centos-7/dca3184/ : SUCCESS in 8m 43s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/74/664574/7/check/openstack-ansible-functional-ubuntu-xenial/1e5cc6b/ : SUCCESS in 7m 15s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/74/664574/7/check/openstack-ansible-lxc-dir-centos-7/d35926f/ : SUCCESS in 17m 53s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/74/664574/7/check/openstack-ansible-lxc-dir-ubuntu-xenial/63c701d/ : SUCCESS in 10m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/74/664574/7/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/677e0c9/ : SUCCESS in 11m 23s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/74/664574/7/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/f26edc4/ : SUCCESS in 9m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/74/664574/7/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/66f9a93/ : SUCCESS in 10m 24s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/74/664574/7/check/openstack-ansible-lxc-zfs-ubuntu-xenial/0b5d24e/ : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":7},{"id":"6e981e8ec3d2b242151e65523c36b1e98a74786b","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-12 17:29:57.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"242370b4aa5f42230cb0e8178f4cc2edc0bf30f6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-12 21:20:28.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f9eca88d05ab32cde43d3d23c930bb043ed0d985","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 21:20:41.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"83282f97b979647abdcd87c43001ede3c39ae36d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 21:59:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/664574/7/gate/build-openstack-sphinx-docs/aadbd5b/html/ : SUCCESS in 3m 07s\n- openstack-ansible-linters http://logs.openstack.org/74/664574/7/gate/openstack-ansible-linters/ba87ce4/ : SUCCESS in 5m 40s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/664574/7/gate/openstack-ansible-functional-centos-7/c4e62fd/ : SUCCESS in 10m 13s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/74/664574/7/gate/openstack-ansible-functional-ubuntu-xenial/2d392da/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":7},{"id":"7c4904c3d10d6a3eadcaefbcb8a9d5f2c035d9fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 21:59:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c8095fc486252cae2b401e9ca2fe23b5f7d72dde","revisions":{"17cc1d2e8bc15ac3b39704cf90b09b6810b59c9d":{"kind":"REWORK","_number":1,"created":"2019-06-11 12:53:24.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/664574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/74/664574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/1"}}},"commit":{"parents":[{"commit":"7f3f5473e0b1f36c0d720b14a5cdc8568d442d67","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7f3f5473e0b1f36c0d720b14a5cdc8568d442d67"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 12:43:55.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 12:43:55.000000000","tz":0},"subject":"Dynamically use lxc 2/3 syntax","message":"Dynamically use lxc 2/3 syntax\n\nChange-Id: I8da5de0074b5ffe470a585bb3ae5394a1dae9efa\nSigned-off-by: Nicolas Bock \u003cnicolasbock@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/17cc1d2e8bc15ac3b39704cf90b09b6810b59c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/17cc1d2e8bc15ac3b39704cf90b09b6810b59c9d"}]},"branch":"refs/heads/stable/queens"},"0dae3b3a5b36035e0fd5b1f777079a0333c596ad":{"kind":"REWORK","_number":2,"created":"2019-06-11 15:28:25.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/664574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/74/664574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/2"}}},"commit":{"parents":[{"commit":"7f3f5473e0b1f36c0d720b14a5cdc8568d442d67","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7f3f5473e0b1f36c0d720b14a5cdc8568d442d67"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 12:43:55.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 15:28:14.000000000","tz":0},"subject":"Dynamically use lxc 2/3 syntax","message":"Dynamically use lxc 2/3 syntax\n\nChange-Id: I8da5de0074b5ffe470a585bb3ae5394a1dae9efa\nSigned-off-by: Nicolas Bock \u003cnicolasbock@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0dae3b3a5b36035e0fd5b1f777079a0333c596ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0dae3b3a5b36035e0fd5b1f777079a0333c596ad"}]},"branch":"refs/heads/stable/queens"},"92fa3226b97eb5dceeb1223ad81319f80b8c08e4":{"kind":"REWORK","_number":3,"created":"2019-06-11 17:23:36.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/664574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/74/664574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/3"}}},"commit":{"parents":[{"commit":"7f3f5473e0b1f36c0d720b14a5cdc8568d442d67","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7f3f5473e0b1f36c0d720b14a5cdc8568d442d67"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 12:43:55.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 17:23:26.000000000","tz":0},"subject":"Dynamically use lxc 2/3 syntax","message":"Dynamically use lxc 2/3 syntax\n\nChange-Id: I8da5de0074b5ffe470a585bb3ae5394a1dae9efa\nSigned-off-by: Nicolas Bock \u003cnicolasbock@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/92fa3226b97eb5dceeb1223ad81319f80b8c08e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/92fa3226b97eb5dceeb1223ad81319f80b8c08e4"}]},"branch":"refs/heads/stable/queens"},"16f2d3c13488696f2ab465dd14d165ebf021a549":{"kind":"REWORK","_number":4,"created":"2019-06-11 17:49:28.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/664574/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/74/664574/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/4"}}},"commit":{"parents":[{"commit":"7f3f5473e0b1f36c0d720b14a5cdc8568d442d67","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7f3f5473e0b1f36c0d720b14a5cdc8568d442d67"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 12:43:55.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 17:48:23.000000000","tz":0},"subject":"Dynamically use lxc 2/3 syntax","message":"Dynamically use lxc 2/3 syntax\n\nChange-Id: I8da5de0074b5ffe470a585bb3ae5394a1dae9efa\nSigned-off-by: Nicolas Bock \u003cnicolasbock@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/16f2d3c13488696f2ab465dd14d165ebf021a549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/16f2d3c13488696f2ab465dd14d165ebf021a549"}]},"branch":"refs/heads/stable/queens"},"cbab905bb3265769d2f087d17fbbf8ad606d21e8":{"kind":"REWORK","_number":5,"created":"2019-06-11 21:18:19.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/664574/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/74/664574/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/5"}}},"commit":{"parents":[{"commit":"7f3f5473e0b1f36c0d720b14a5cdc8568d442d67","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7f3f5473e0b1f36c0d720b14a5cdc8568d442d67"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 12:43:55.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 21:18:09.000000000","tz":0},"subject":"Dynamically use lxc 2/3 syntax","message":"Dynamically use lxc 2/3 syntax\n\nChange-Id: I8da5de0074b5ffe470a585bb3ae5394a1dae9efa\nSigned-off-by: Nicolas Bock \u003cnicolasbock@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cbab905bb3265769d2f087d17fbbf8ad606d21e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cbab905bb3265769d2f087d17fbbf8ad606d21e8"}]},"branch":"refs/heads/stable/queens"},"f8ab32ad5c26ce1d46646e7391c4b5e5ad3701e5":{"kind":"REWORK","_number":6,"created":"2019-06-11 22:13:18.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/664574/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/74/664574/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/6"}}},"commit":{"parents":[{"commit":"7f3f5473e0b1f36c0d720b14a5cdc8568d442d67","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7f3f5473e0b1f36c0d720b14a5cdc8568d442d67"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 12:43:55.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 22:13:07.000000000","tz":0},"subject":"Dynamically use lxc 2/3 syntax","message":"Dynamically use lxc 2/3 syntax\n\nChange-Id: I8da5de0074b5ffe470a585bb3ae5394a1dae9efa\nSigned-off-by: Nicolas Bock \u003cnicolasbock@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f8ab32ad5c26ce1d46646e7391c4b5e5ad3701e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f8ab32ad5c26ce1d46646e7391c4b5e5ad3701e5"}]},"branch":"refs/heads/stable/queens"},"c8095fc486252cae2b401e9ca2fe23b5f7d72dde":{"kind":"REWORK","_number":7,"created":"2019-06-11 22:29:25.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/664574/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/74/664574/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/74/664574/7"}}},"commit":{"parents":[{"commit":"7f3f5473e0b1f36c0d720b14a5cdc8568d442d67","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7f3f5473e0b1f36c0d720b14a5cdc8568d442d67"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 12:43:55.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-06-11 22:29:14.000000000","tz":0},"subject":"Dynamically use lxc 2/3 syntax","message":"Dynamically use lxc 2/3 syntax\n\nChange-Id: I8da5de0074b5ffe470a585bb3ae5394a1dae9efa\nSigned-off-by: Nicolas Bock \u003cnicolasbock@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c8095fc486252cae2b401e9ca2fe23b5f7d72dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c8095fc486252cae2b401e9ca2fe23b5f7d72dde"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
