)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~667102","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I3ad0d46d5eb22ae43fd760a7d701f4d9f1dc9a9e","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I3ad0d46d5eb22ae43fd760a7d701f4d9f1dc9a9e","subject":"Update LXC base image to 18.04.2","status":"MERGED","created":"2019-06-24 13:40:11.000000000","updated":"2019-07-10 16:31:58.000000000","submitted":"2019-07-10 16:31:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667102-1562776318093-0a1ee91c","meta_rev_id":"0c39b91ce7203c9578760100f08129a757b35bff","_number":667102,"virtual_id_number":667102,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-10 16:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-24 14:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-07-10 16:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-07-10 16:03:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 14:33:49.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-07-10 16:03:52.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 16:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"701de9da12deadcd05327317d0bf71061d40d1f1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-24 13:40:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9ca44a430366a6e936407e2c9896d98410e6762","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-24 14:33:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"219d28fd7f4902ef0a1140988303b3a5fed2cbb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:51:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/02/667102/1/check/openstack-ansible-linters/bf0dfe3/ : SUCCESS in 5m 43s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/02/667102/1/check/openstack-ansible-functional-debian-stable/9939a2f/ : SUCCESS in 9m 10s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/02/667102/1/check/openstack-ansible-functional-centos-7/4489fec/ : SUCCESS in 9m 46s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/02/667102/1/check/openstack-ansible-functional-ubuntu-bionic/01afeba/ : SUCCESS in 7m 56s\n- openstack-tox-docs http://logs.openstack.org/02/667102/1/check/openstack-tox-docs/b7a34a4/html/ : SUCCESS in 3m 05s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-dir-centos-7/77fe151/ : SUCCESS in 18m 51s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-dir-opensuse-150/ee54973/ : FAILURE in 20m 47s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/1bdbe8b/ : SUCCESS in 14m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/3ee94e5/ : SUCCESS in 11m 23s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-btrfs-opensuse-150/bf5dbdf/ : FAILURE in 14m 44s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/155a595/ : SUCCESS in 20m 51s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-machinectl-centos-7/4cc19e7/ : SUCCESS in 13m 19s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-machinectl-opensuse-150/81b8549/ : FAILURE in 16m 01s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/28135ad/ : SUCCESS in 11m 40s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/3c23508/ : FAILURE in 14m 42s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/02/667102/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/360c11f/ : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0d8df07887862ce31233299376a10534c5a49e20","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 16:03:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbd1875d120f81f1f7234b38877500f5d641758d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:04:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fc9b29502487a224e66afd182e5df9db86ed97f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:31:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/02/667102/1/gate/openstack-ansible-linters/4c84653/ : SUCCESS in 7m 01s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/02/667102/1/gate/openstack-ansible-functional-debian-stable/732755d/ : SUCCESS in 17m 56s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/02/667102/1/gate/openstack-ansible-functional-centos-7/c090e9c/ : SUCCESS in 10m 56s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/02/667102/1/gate/openstack-ansible-functional-ubuntu-bionic/fac8544/ : SUCCESS in 7m 40s\n- openstack-tox-docs http://logs.openstack.org/02/667102/1/gate/openstack-tox-docs/e70cb21/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0c39b91ce7203c9578760100f08129a757b35bff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:31:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b075f214ebf834bfd1efd4c6e853a45a501b4792","revisions":{"b075f214ebf834bfd1efd4c6e853a45a501b4792":{"kind":"REWORK","_number":1,"created":"2019-06-24 13:40:11.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/02/667102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/02/667102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/02/667102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/02/667102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/02/667102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/02/667102/1"}}},"commit":{"parents":[{"commit":"a6cae27fa3e6d03b48ba34468df4af90c77f4880","subject":"Merge \"Revert \"Temporary fix of failing aria2 on centos\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6cae27fa3e6d03b48ba34468df4af90c77f4880"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-24 13:39:06.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-24 13:39:06.000000000","tz":60},"subject":"Update LXC base image to 18.04.2","message":"Update LXC base image to 18.04.2\n\nChange-Id: I3ad0d46d5eb22ae43fd760a7d701f4d9f1dc9a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b075f214ebf834bfd1efd4c6e853a45a501b4792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b075f214ebf834bfd1efd4c6e853a45a501b4792"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
