)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~670126","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fstein~Ibe4112ef947513439643025ddf0fd543b28aba70","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/stein","hashtags":[],"change_id":"Ibe4112ef947513439643025ddf0fd543b28aba70","subject":"Fix copying apt keys from host to containers","status":"MERGED","created":"2019-07-10 16:43:05.000000000","updated":"2019-07-23 15:55:05.000000000","submitted":"2019-07-23 15:55:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670126-1563897305256-7453feb3","meta_rev_id":"704556ed9dc9fe89111be3bfa4e9a1b45d490dcf","_number":670126,"virtual_id_number":670126,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-23 15:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"date":"2019-07-23 10:30:29.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":1,"date":"2019-07-13 18:52:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-23 07:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-12 22:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-15 22:39:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":1,"date":"2019-07-13 09:06:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-07-23 07:43:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 22:56:57.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-13 09:06:11.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-13 18:52:13.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-15 22:39:52.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-07-23 07:43:10.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-23 15:55:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c40105820d9678565fd1f25b40302c5cb8364808","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-10 16:43:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"24d4d5426860b04a93a00010c518fe54503a017f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-10 16:43:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 70978b16fe5b66504594c22a43f5d463541635d8","accounts_in_message":[],"_revision_number":1},{"id":"628eaf681d89f860cd7076804e838b6e2e789190","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:57:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/26/670126/1/check/openstack-ansible-linters/27eab0e/ : SUCCESS in 6m 03s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-debian-stable/47e8d50/ : SUCCESS in 8m 46s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-centos-7/f573428/ : SUCCESS in 9m 03s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-ubuntu-bionic/4e1ca56/ : SUCCESS in 12m 19s\n- openstack-tox-docs http://logs.openstack.org/26/670126/1/check/openstack-tox-docs/592ba1f/html/ : SUCCESS in 3m 45s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-centos-7/f48daaf/ : SUCCESS in 18m 41s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-opensuse-150/d2fe88a/ : FAILURE in 17m 35s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/2b850c1/ : SUCCESS in 16m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/6af3060/ : SUCCESS in 14m 44s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-opensuse-150/f5b9623/ : FAILURE in 13m 13s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/e5f8361/ : SUCCESS in 11m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-centos-7/69ee3f0/ : SUCCESS in 15m 14s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-opensuse-150/08f2a89/ : FAILURE in 12m 39s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/6bcdd35/ : SUCCESS in 17m 22s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/1dfeff7/ : FAILURE in 19m 16s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/a5151a1/ : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"64ec83bdb9bf348983e112b6bf37fe117f6ddac0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-12 22:56:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"defdf55ca9f669f62f495c37b5ae4be5f67d34f4","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-13 09:06:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0c46b1e8b193d43d00afc3461ed8901db985dba3","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 18:52:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3e1b77266f5dfc9abedc5034c4bbdfc35e83a1ad","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-07-15 22:39:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"46e77b811d5ab3e74e9cafa8da03524dc9a95e00","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 07:43:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10150c9156fbdbbd5c00c7f659c392f33eebdbdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:43:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aadf51203762c642abbf5e587982401a4c0cf30f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:59:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/26/670126/1/gate/openstack-ansible-linters/e1ef486/ : SUCCESS in 5m 50s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/26/670126/1/gate/openstack-ansible-functional-debian-stable/e04b1cf/ : SUCCESS in 9m 22s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/26/670126/1/gate/openstack-ansible-functional-centos-7/a2fda1c/ : FAILURE in 7m 16s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/26/670126/1/gate/openstack-ansible-functional-ubuntu-bionic/b0443ba/ : SUCCESS in 9m 51s\n- openstack-tox-docs http://logs.openstack.org/26/670126/1/gate/openstack-tox-docs/59aa43e/html/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"43ba17389c694b40ae892b6a98613b0ad06bfd29","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 08:07:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9dcff1b7a721862b9c601b4b4f889afb3ff43517","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:38:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/26/670126/1/check/openstack-ansible-linters/e94fd11/ : FAILURE in 6m 17s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-debian-stable/05b77ed/ : SUCCESS in 11m 43s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-centos-7/2f9d57b/ : SUCCESS in 8m 23s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-ubuntu-bionic/3621ac2/ : SUCCESS in 9m 13s\n- openstack-tox-docs http://logs.openstack.org/26/670126/1/check/openstack-tox-docs/2e72fe5/html/ : SUCCESS in 3m 21s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-centos-7/20221a1/ : SUCCESS in 15m 53s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-opensuse-150/f65e2c0/ : FAILURE in 13m 54s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/269cb72/ : SUCCESS in 12m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/e3412a8/ : SUCCESS in 13m 36s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-opensuse-150/c338178/ : FAILURE in 21m 32s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/55c2e98/ : SUCCESS in 12m 37s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-centos-7/780939e/ : SUCCESS in 14m 46s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-opensuse-150/b39235a/ : FAILURE in 14m 48s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/b1c4b08/ : SUCCESS in 14m 59s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/96a92dd/ : FAILURE in 15m 20s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/1cdc5ce/ : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"dbfa4b46e63304f02a8ed56bc49d7f64b84db240","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 08:41:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"02d82b0d081e5517575dc07ced1943b7bf79d469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:10:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/26/670126/1/check/openstack-ansible-linters/e6b35ee/ : FAILURE in 7m 34s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-debian-stable/069b282/ : SUCCESS in 9m 04s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-centos-7/d920ff7/ : SUCCESS in 9m 06s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-ubuntu-bionic/d4faa0d/ : SUCCESS in 13m 55s\n- openstack-tox-docs http://logs.openstack.org/26/670126/1/check/openstack-tox-docs/fe18196/html/ : SUCCESS in 3m 32s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-centos-7/6d002a2/ : SUCCESS in 15m 36s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-opensuse-150/b8ec8cb/ : FAILURE in 8m 25s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/732b24f/ : SUCCESS in 12m 03s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/caa061a/ : FAILURE in 5m 56s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-opensuse-150/5456cd7/ : FAILURE in 8m 45s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/9a6893d/ : SUCCESS in 11m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-centos-7/07b5e7c/ : SUCCESS in 14m 31s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-opensuse-150/e045443/ : FAILURE in 15m 46s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/f3d85ad/ : FAILURE in 7m 19s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/03d1343/ : FAILURE in 17m 55s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/4311b61/ : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"81672971ffb70df2860f443e53bdc6a97b53cecc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-23 10:30:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eed8a4aa764f5a613f228e9bb71f8445596d5442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 11:04:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/26/670126/1/check/openstack-ansible-linters/6c7b01a/ : SUCCESS in 5m 51s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-debian-stable/be8ec09/ : SUCCESS in 11m 08s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-centos-7/9b9153d/ : SUCCESS in 11m 24s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-ubuntu-bionic/767e800/ : SUCCESS in 8m 04s\n- openstack-tox-docs http://logs.openstack.org/26/670126/1/check/openstack-tox-docs/fa1ebfd/html/ : SUCCESS in 3m 17s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-centos-7/edb063f/ : FAILURE in 6m 33s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-opensuse-150/61ac9e5/ : FAILURE in 13m 48s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/76e2766/ : SUCCESS in 12m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/0da8823/ : SUCCESS in 12m 20s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-opensuse-150/f335621/ : FAILURE in 14m 51s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/415fc17/ : SUCCESS in 11m 31s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-centos-7/b7d397a/ : SUCCESS in 15m 04s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-opensuse-150/4727627/ : FAILURE in 16m 05s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/6b493bb/ : SUCCESS in 14m 32s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/1483fd6/ : FAILURE in 23m 13s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/45a7dd8/ : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"df7dac1a1e23b8fc99fc013eb3db712c7db5b1ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 11:58:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"87964292aba773721aa330791b4c9a7e25b288a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:29:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/26/670126/1/check/openstack-ansible-linters/454cadd/ : SUCCESS in 7m 15s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-debian-stable/2413d9e/ : SUCCESS in 8m 20s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-centos-7/f074162/ : SUCCESS in 10m 03s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-ubuntu-bionic/f137463/ : SUCCESS in 10m 30s\n- openstack-tox-docs http://logs.openstack.org/26/670126/1/check/openstack-tox-docs/e55c1d3/html/ : SUCCESS in 3m 08s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-centos-7/7d14474/ : SUCCESS in 15m 34s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-opensuse-150/5135757/ : FAILURE in 16m 37s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/05246e9/ : SUCCESS in 11m 44s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/de4a4ee/ : SUCCESS in 16m 19s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-opensuse-150/53d7fb5/ : FAILURE in 14m 00s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/0e49398/ : SUCCESS in 12m 50s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-centos-7/133b902/ : SUCCESS in 15m 19s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-opensuse-150/7610a18/ : FAILURE in 13m 48s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/530b990/ : FAILURE in 7m 14s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/fdb1d69/ : FAILURE in 16m 53s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/af02449/ : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e0ad69869e9134adbc41500aede523c966975787","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-23 15:05:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"110700f6fd7dc25406ee8b65ecabbb5470672801","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:38:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/26/670126/1/check/openstack-ansible-linters/76f5764/ : SUCCESS in 6m 36s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-debian-stable/1b8f000/ : SUCCESS in 10m 15s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-centos-7/0459055/ : SUCCESS in 14m 36s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-functional-ubuntu-bionic/591e2c1/ : SUCCESS in 14m 44s\n- openstack-tox-docs http://logs.openstack.org/26/670126/1/check/openstack-tox-docs/04429cd/html/ : SUCCESS in 2m 58s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-centos-7/3db2246/ : SUCCESS in 15m 24s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-opensuse-150/ad2a853/ : FAILURE in 15m 03s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/ca9271c/ : SUCCESS in 19m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/8c6507a/ : SUCCESS in 19m 13s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-btrfs-opensuse-150/bedb6e5/ : FAILURE in 24m 19s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/d68753b/ : SUCCESS in 14m 25s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-centos-7/17bcd7d/ : SUCCESS in 15m 27s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-machinectl-opensuse-150/ca19842/ : FAILURE in 20m 23s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/c05186b/ : SUCCESS in 13m 46s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/f75c72d/ : FAILURE in 20m 46s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/26/670126/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/7fdc076/ : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"576fa0f1efc0b5a6f31d35a4a96fcf5bff5a48b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:38:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6690869c1e407a562a9deddc57790697c44bea52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:55:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"704556ed9dc9fe89111be3bfa4e9a1b45d490dcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:55:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/26/670126/1/gate/openstack-ansible-linters/b9f0bac/ : SUCCESS in 7m 56s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/26/670126/1/gate/openstack-ansible-functional-debian-stable/fe5505b/ : SUCCESS in 9m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/26/670126/1/gate/openstack-ansible-functional-centos-7/f7a6d9a/ : SUCCESS in 10m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/26/670126/1/gate/openstack-ansible-functional-ubuntu-bionic/0d97fae/ : SUCCESS in 9m 36s\n- openstack-tox-docs http://logs.openstack.org/26/670126/1/gate/openstack-tox-docs/0922a46/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32ede70043039fca1e5ac2e966d07d827e43a14f","revisions":{"32ede70043039fca1e5ac2e966d07d827e43a14f":{"kind":"REWORK","_number":1,"created":"2019-07-10 16:43:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/26/670126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/26/670126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/26/670126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/26/670126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/26/670126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/26/670126/1"}}},"commit":{"parents":[{"commit":"1b402c9b1b36c084fceb1a63a8a12acd73ade4d3","subject":"Revert \"Temporary fix of failing aria2 on centos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b402c9b1b36c084fceb1a63a8a12acd73ade4d3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-24 13:30:07.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-10 16:43:05.000000000","tz":0},"subject":"Fix copying apt keys from host to containers","message":"Fix copying apt keys from host to containers\n\nWithout the trailing slash the keys had an extra level of directories\nas follows: /etc/apt/trusted.gpg.d/trusted.gpg.d/\u003ckeys\u003e\n\nChange-Id: Ibe4112ef947513439643025ddf0fd543b28aba70\n(cherry picked from commit 9fd34ada152889c06c744ea5b4928c3554c9b4c5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/32ede70043039fca1e5ac2e966d07d827e43a14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/32ede70043039fca1e5ac2e966d07d827e43a14f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
