)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~670127","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Frocky~Ibe4112ef947513439643025ddf0fd543b28aba70","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/rocky","hashtags":[],"change_id":"Ibe4112ef947513439643025ddf0fd543b28aba70","subject":"Fix copying apt keys from host to containers","status":"MERGED","created":"2019-07-10 16:43:15.000000000","updated":"2019-07-23 08:19:20.000000000","submitted":"2019-07-23 08:19:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670127-1563869960889-a8feb72e","meta_rev_id":"60aebc39effe0486e0200da3fec2413241299e1d","_number":670127,"virtual_id_number":670127,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-23 08:19:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":1,"date":"2019-07-13 18:52:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-23 07:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-12 22:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-13 09:05:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-07-23 07:44:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 22:57:01.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-13 09:05:48.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-13 18:52:27.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-23 07:44:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-23 08:19:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"517bf0a460d44958f266a533cb7e1fa46468716a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-10 16:43:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"d0ae9d685000083058f18ede830c06c57716e066","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:38:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/27/670127/1/check/openstack-ansible-linters/73906b3/ : SUCCESS in 6m 00s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/27/670127/1/check/openstack-ansible-functional-centos-7/c5af755/ : SUCCESS in 9m 11s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/27/670127/1/check/openstack-ansible-functional-opensuse-423/f9cf3e5/ : FAILURE in 7m 33s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/27/670127/1/check/openstack-ansible-functional-opensuse-150/ef50c1c/ : FAILURE in 7m 28s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/27/670127/1/check/openstack-ansible-functional-ubuntu-bionic/ce74d3c/ : SUCCESS in 7m 15s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/27/670127/1/check/openstack-ansible-functional-ubuntu-xenial/9c57401/ : SUCCESS in 9m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/27/670127/1/check/build-openstack-sphinx-docs/979525c/html/ : SUCCESS in 3m 25s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-dir-centos-7/b630f33/ : SUCCESS in 14m 17s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-dir-opensuse-423/980ae60/ : FAILURE in 9m 48s (non-voting)\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-dir-opensuse-150/94ad33e/ : FAILURE in 12m 01s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-dir-ubuntu-xenial/6e143ef/ : SUCCESS in 12m 05s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/ee8540e/ : SUCCESS in 11m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/c0dcd1e/ : SUCCESS in 11m 20s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/b1d426b/ : SUCCESS in 10m 46s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-btrfs-centos-7/c60ec46/ : SUCCESS in 15m 57s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-btrfs-opensuse-423/e154e76/ : FAILURE in 11m 42s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-btrfs-opensuse-150/8c61b25/ : FAILURE in 12m 06s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/b6ea18b/ : SUCCESS in 14m 49s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/ab723c1/ : FAILURE in 10m 46s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-machinectl-centos-7/782180f/ : SUCCESS in 12m 33s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-machinectl-opensuse-423/bb100a9/ : FAILURE in 11m 16s (non-voting)\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-machinectl-opensuse-150/129ada9/ : FAILURE in 11m 48s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/25732e1/ : SUCCESS in 14m 49s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/92fccf5/ : SUCCESS in 12m 01s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-overlayfs-centos-7/7649f9a/ : SUCCESS in 16m 38s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-overlayfs-opensuse-423/5231dca/ : FAILURE in 7m 22s (non-voting)\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/c631d96/ : FAILURE in 8m 31s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-zfs-ubuntu-xenial/5725563/ : SUCCESS in 13m 05s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/27/670127/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/9db8937/ : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d1df5e0ab7ccb87120934c38341418e26fa227ca","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-12 22:57:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"071781dcb82a1db8c116824454f42aa7e8c0d340","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-13 09:05:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b622c27f96a2fed278e58aaaf108f5735a3ae2be","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 18:52:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"94d498ed265e05c07876b9053831343aba545a1b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 07:44:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32a8daf9f7cd3161984a1c24a13dcb5d4f21c431","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:44:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c85875b7cf10e10f2f2bcb524d1a5677338c710b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:19:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/27/670127/1/gate/openstack-ansible-linters/a926330/ : SUCCESS in 5m 36s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/27/670127/1/gate/openstack-ansible-functional-centos-7/92a0aa2/ : SUCCESS in 9m 12s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/27/670127/1/gate/openstack-ansible-functional-opensuse-423/7fb8fbd/ : FAILURE in 7m 06s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/27/670127/1/gate/openstack-ansible-functional-ubuntu-bionic/96d1de4/ : SUCCESS in 10m 05s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/27/670127/1/gate/openstack-ansible-functional-ubuntu-xenial/08cbdb8/ : SUCCESS in 10m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/27/670127/1/gate/build-openstack-sphinx-docs/92d6b26/html/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"60aebc39effe0486e0200da3fec2413241299e1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:19:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70978b16fe5b66504594c22a43f5d463541635d8","revisions":{"70978b16fe5b66504594c22a43f5d463541635d8":{"kind":"REWORK","_number":1,"created":"2019-07-10 16:43:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/27/670127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/27/670127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/27/670127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/27/670127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/27/670127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/27/670127/1"}}},"commit":{"parents":[{"commit":"fcd735935b881475cf627ae7264d2ebaf7554ede","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fcd735935b881475cf627ae7264d2ebaf7554ede"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-24 13:30:07.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-10 16:43:15.000000000","tz":0},"subject":"Fix copying apt keys from host to containers","message":"Fix copying apt keys from host to containers\n\nWithout the trailing slash the keys had an extra level of directories\nas follows: /etc/apt/trusted.gpg.d/trusted.gpg.d/\u003ckeys\u003e\n\nChange-Id: Ibe4112ef947513439643025ddf0fd543b28aba70\n(cherry picked from commit 9fd34ada152889c06c744ea5b4928c3554c9b4c5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/70978b16fe5b66504594c22a43f5d463541635d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/70978b16fe5b66504594c22a43f5d463541635d8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
