)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~671583","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I6f880f9571afa2a9224ab96dba62aba10d9adafa","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I6f880f9571afa2a9224ab96dba62aba10d9adafa","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-18 22:16:52.000000000","updated":"2019-07-19 12:40:59.000000000","submitted":"2019-07-19 12:40:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671583-1563540059438-22ab38c7","meta_rev_id":"f88bdb057a3f169c67af27bcae56ee3b4160872e","_number":671583,"virtual_id_number":671583,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 12:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-19 12:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-19 12:23:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 12:23:59.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 12:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23d2ed3f9c6aa04cfe7922c9c2ad1d11cf64cbb4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-18 22:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28b9c612ae24bdec6e108f3c423bac34a67e1dad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 22:47:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/83/671583/1/check/openstack-ansible-linters/0d8fbd9/ : SUCCESS in 6m 39s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/671583/1/check/openstack-ansible-functional-debian-stable/825d942/ : SUCCESS in 8m 13s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/671583/1/check/openstack-ansible-functional-centos-7/47d86a4/ : SUCCESS in 10m 29s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/671583/1/check/openstack-ansible-functional-ubuntu-bionic/133a4c4/ : SUCCESS in 9m 41s\n- openstack-tox-docs http://logs.openstack.org/83/671583/1/check/openstack-tox-docs/7c9c451/html/ : SUCCESS in 4m 03s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-dir-centos-7/be145a8/ : SUCCESS in 15m 28s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-dir-opensuse-150/5878799/ : FAILURE in 18m 12s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/0e9b702/ : SUCCESS in 12m 43s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/7ac9be0/ : SUCCESS in 12m 54s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-btrfs-opensuse-150/5173cdc/ : FAILURE in 22m 03s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/14f1ac9/ : SUCCESS in 11m 15s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-machinectl-centos-7/329e7c4/ : SUCCESS in 13m 42s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-machinectl-opensuse-150/30d78ac/ : FAILURE in 12m 34s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/74e5892/ : SUCCESS in 12m 14s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/d2ca6f3/ : FAILURE in 17m 00s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/83/671583/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/23ed770/ : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2951a9a9050b11ef63d0169e8eafd031b41c5312","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 12:23:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ceb96c75c9a562deb22c057ec721273c6f89518","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:24:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13d765c2288e693d155595b9775180b325a97a0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:40:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1499513dde9777b24b6a245925223fba16ac49ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:40:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/83/671583/1/gate/openstack-ansible-linters/403834a/ : SUCCESS in 7m 27s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/671583/1/gate/openstack-ansible-functional-debian-stable/fabc826/ : SUCCESS in 7m 37s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/671583/1/gate/openstack-ansible-functional-centos-7/f7a2eaf/ : SUCCESS in 11m 24s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/671583/1/gate/openstack-ansible-functional-ubuntu-bionic/a53b659/ : SUCCESS in 7m 42s\n- openstack-tox-docs http://logs.openstack.org/83/671583/1/gate/openstack-tox-docs/509be45/html/ : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b307cebe6f6d0c75bca0ba5c884156d1971f1651","revisions":{"b307cebe6f6d0c75bca0ba5c884156d1971f1651":{"kind":"REWORK","_number":1,"created":"2019-07-18 22:16:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/671583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/83/671583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/671583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/671583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/671583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/671583/1"}}},"commit":{"parents":[{"commit":"9b9a16a35e905e7e38c755b479319ccfa58dd500","subject":"Merge \"Update LXC base image to 18.04.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9b9a16a35e905e7e38c755b479319ccfa58dd500"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:16:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:16:50.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I6f880f9571afa2a9224ab96dba62aba10d9adafa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b307cebe6f6d0c75bca0ba5c884156d1971f1651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b307cebe6f6d0c75bca0ba5c884156d1971f1651"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
