)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~677988","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Frocky~Ic5083ab9b2c14e528f8a41415279ff74a3511313","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/rocky","topic":"goodbye_42.3","hashtags":[],"change_id":"Ic5083ab9b2c14e528f8a41415279ff74a3511313","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-08-22 11:53:38.000000000","updated":"2019-08-27 13:29:17.000000000","submitted":"2019-08-27 13:29:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677988-1566912557438-6ccbb2a0","meta_rev_id":"6b37b001cf1e10ffc23880bf815f6986fa7650d3","_number":677988,"virtual_id_number":677988,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-27 13:29:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2019-08-27 08:46:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-27 12:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-27 13:01:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-08-27 13:01:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 08:46:53.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-27 12:29:39.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-27 13:01:59.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-27 13:29:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d4a7d4b1d1c82fb0b950935b23c9cf9d367f64b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-22 11:53:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"567ffbf1f2084f18493dbf82b3e696a301f3c327","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:53:56.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-ansible-lxc_hosts on branch stable/rocky.\nThe error was:\n\n  Job openstack-ansible-lxc-btrfs-opensuse-423 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-ansible-role-jobs\n        - check-requirements\n        - publish-openstack-sphinx-docs\n        - release-notes-jobs\n      check:\n        jobs:\n          - openstack-ansible-lxc-dir-centos-7\n          - openstack-ansible-lxc-dir-opensuse-423\n          - openstack-ansible-lxc-dir-opensuse-150\n          - openstack-ansible-lxc-dir-ubuntu-xenial\n          - openstack-ansible-lxc-dir-ubuntu-bionic\n          - openstack-ansible-lxc-btrfs-ubuntu-xenial\n          - openstack-ansible-lxc-btrfs-ubuntu-bionic\n          - openstack-ansible-lxc-btrfs-centos-7\n          - openstack-ansible-lxc-btrfs-opensuse-423\n          - openstack-ansible-lxc-btrfs-opensuse-150\n          - openstack-ansible-lxc-machinectl-ubuntu-xenial\n          - openstack-ansible-lxc-machinectl-ubuntu-bionic\n          - openstack-ansible-lxc-machinectl-centos-7\n          - openstack-ansible-lxc-machinectl-opensuse-423\n          - openstack-ansible-lxc-machinectl-opensuse-150\n          - openstack-ansible-lxc-overlayfs-ubuntu-xenial\n          - openstack-ansible-lxc-overlayfs-ubuntu-bionic\n          - openstack-ansible-lxc-overlayfs-centos-7\n          - openstack-ansible-lxc-overlayfs-opensuse-423\n          - openstack-ansible-lxc-overlayfs-opensuse-150\n          - openstack-ansible-lxc-zfs-ubuntu-xenial\n          - openstack-ansible-lxc-zfs-ubuntu-bionic\n\n  in \"openstack/openstack-ansible-lxc_hosts/zuul.d/project.yaml@stable/rocky\", line 15, column 3","accounts_in_message":[],"_revision_number":1},{"id":"a72534fa98c290e00b24910d9f15feeef7e7f36b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-22 11:57:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"67e890367cc648d258f8d2b3932be8fffbe1b57a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 12:53:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c6def1854d014508af2ab4a6ec5ceff3 : SUCCESS in 6m 04s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/fa16e01a3fc04ca3af1ae893d179e50b : POST_FAILURE in 9m 50s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/a04d6f8d85c84eb6b3d689de929db3cc : POST_FAILURE in 8m 36s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/607926e4d9924c3a820a3aefd4ab1b6b : SUCCESS in 10m 24s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/50e2a182220e48b9a77418f17c5a84f0 : SUCCESS in 11m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1007889bd32e4e4c98a73b37b38619a1 : SUCCESS in 5m 44s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/58b1615624364c1e99802fcd89264636 : POST_FAILURE in 14m 46s\n- openstack-ansible-lxc-dir-opensuse-150 https://zuul.opendev.org/t/openstack/build/1b1a17dfdeaf4fdaa61ca88c910dcab7 : POST_FAILURE in 9m 42s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/505aef63db294a8a98a4672eb6d2867b : SUCCESS in 14m 58s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/441b220b12ff4d7ebbcfdc30e49e0615 : SUCCESS in 19m 30s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/28836c8f12b6482ca1926cc4b4ba44b5 : SUCCESS in 13m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3698999545bc402bbc1783649fa3527d : SUCCESS in 13m 21s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 https://zuul.opendev.org/t/openstack/build/90bdc15fe815443db231e38725858842 : POST_FAILURE in 14m 13s\n- openstack-ansible-lxc-btrfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/1b4504314e864b6c914c2dab5af9085e : POST_FAILURE in 8m 27s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e7dda35a83e743919f6d3f06957e3d02 : SUCCESS in 21m 23s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d758c86796c48b3bfcf27443c49667d : FAILURE in 7m 52s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/22347483b8ac40358b7a90ffbaadc4eb : POST_FAILURE in 14m 20s\n- openstack-ansible-lxc-machinectl-opensuse-150 https://zuul.opendev.org/t/openstack/build/5c19846478874172b215f31cc56ce2ad : POST_FAILURE in 12m 06s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/571f60f1f37c43969ef41bdc1bd59103 : SUCCESS in 20m 00s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ade4c625039424386507d79b9db0409 : SUCCESS in 17m 11s\n- openstack-ansible-lxc-overlayfs-centos-7 https://zuul.opendev.org/t/openstack/build/997e48c39b934d20b841e10880996002 : POST_FAILURE in 16m 05s\n- openstack-ansible-lxc-overlayfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/7ddab58168434f6bb303cb8c9c9749f4 : POST_FAILURE in 10m 33s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/608fbd7fdd7d4271b7a51759a0002dd5 : SUCCESS in 14m 49s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3d1091af31b414281084e88305efe93 : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"80420dcc5e56f8a95a03aee76ddc9a5408085faa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 08:46:53.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck - post_failure should be fixed now","accounts_in_message":[],"_revision_number":2},{"id":"6ec1ef62094b8307c625e2b636876636a656b89b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 09:27:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5ea55014d86443f8bb6e1e2f9c71b852 : SUCCESS in 8m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/77b50ad83e134a13bd2215935ccad10c : SUCCESS in 12m 17s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/e28bfbaace494eb687b3aed43db2abff : RETRY_LIMIT in 4m 23s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a37910ed7fec47adab84cb5fabc44f15 : SUCCESS in 8m 56s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f83540891e74458393088449bd720169 : SUCCESS in 8m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/74787442eb04497e87654972ff052e72 : SUCCESS in 4m 22s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/acaf7c2dbc174213a92d59e99a3cccec : SUCCESS in 20m 10s\n- openstack-ansible-lxc-dir-opensuse-150 https://zuul.opendev.org/t/openstack/build/c3ed0721e15249d4aaa4b2d7db97c96a : RETRY_LIMIT in 4m 57s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/97c87ea891bc4a398024f446f62f8dcd : SUCCESS in 13m 55s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1e7e68ead464d1cacaef1c46d004851 : SUCCESS in 12m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4bb4239a56ff4b1dbca7b26f3e68a4a6 : SUCCESS in 14m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f1ea315d32946dba5625064b503a4ad : SUCCESS in 12m 45s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 https://zuul.opendev.org/t/openstack/build/d58a2d9dc78843e2b940392bd2f5768e : SUCCESS in 18m 43s\n- openstack-ansible-lxc-btrfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/5ce94ce6a9084bc0a6a57bf3f2d83a8a : RETRY_LIMIT in 7m 32s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/34a80988f5af4f2b92312b6a058ea5ee : SUCCESS in 11m 47s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81fec6066c37475995a9f086b5d1aed8 : FAILURE in 16m 55s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/640dde3ca8dd485499b89885523c9fb9 : SUCCESS in 14m 08s\n- openstack-ansible-lxc-machinectl-opensuse-150 https://zuul.opendev.org/t/openstack/build/9e18021296004939a153ebd2567be21d : RETRY_LIMIT in 4m 06s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4ce9774c7de1408caa9e8b19682fea78 : SUCCESS in 14m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f374cbb285445f5956e8c0145e93505 : SUCCESS in 13m 40s\n- openstack-ansible-lxc-overlayfs-centos-7 https://zuul.opendev.org/t/openstack/build/40340e810a7a4df3a74d54fd7a827162 : SUCCESS in 13m 47s\n- openstack-ansible-lxc-overlayfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/a3fb1124ebcf434a97bdbe82a6dee659 : RETRY_LIMIT in 3m 15s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/15297514da734b0694281ee501601a20 : SUCCESS in 15m 35s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a63b08e1fdc4d649f12c12d9293f43b : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c5779f5ab6e932aa8b7485164381808706d11905","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:29:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1cef6d65bb7684a27db87edac5ef0f5fd1a19fbd","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-27 13:01:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3955dd7c3245ae52ca1fbbcdd36bb1017ec07473","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:02:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64e31ca350511c80cd06b3c8824d5255d03009f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:29:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2f1f30af23cdbce460b0287ab8d261818cea41ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:29:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/93f4c859efba48588b216d2b93f6d720 : SUCCESS in 5m 43s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/76c5a6d3681e47c08734d903ae875b14 : SUCCESS in 13m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/835dc359bb0b444e8a279626a9716ee6 : SUCCESS in 11m 02s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/355283688d80402289f96c4746dfaaf6 : SUCCESS in 11m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/095278dd320044ea99c1d1f0253e273e : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e4edc2b9f2c5f38066381a1f486100795b5a1b8","revisions":{"9ab7a417a38a2b0960fcc5df883e58cb2d4d3469":{"kind":"REWORK","_number":1,"created":"2019-08-22 11:53:38.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/88/677988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/88/677988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/88/677988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/88/677988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/88/677988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/88/677988/1"}}},"commit":{"parents":[{"commit":"70978b16fe5b66504594c22a43f5d463541635d8","subject":"Fix copying apt keys from host to containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/70978b16fe5b66504594c22a43f5d463541635d8"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:53:18.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:53:35.000000000","tz":120},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job, we test with openSUSE 15 now (as\non master). This allows to remove the nodes from CI.\n\nDepends-On: https://review.opendev.org/677947\nChange-Id: Ic5083ab9b2c14e528f8a41415279ff74a3511313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9ab7a417a38a2b0960fcc5df883e58cb2d4d3469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9ab7a417a38a2b0960fcc5df883e58cb2d4d3469"}]},"branch":"refs/heads/stable/rocky"},"4e4edc2b9f2c5f38066381a1f486100795b5a1b8":{"kind":"REWORK","_number":2,"created":"2019-08-22 11:57:56.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/88/677988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/88/677988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/88/677988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/88/677988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/88/677988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/88/677988/2"}}},"commit":{"parents":[{"commit":"70978b16fe5b66504594c22a43f5d463541635d8","subject":"Fix copying apt keys from host to containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/70978b16fe5b66504594c22a43f5d463541635d8"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:53:18.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:57:44.000000000","tz":120},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job, we test with openSUSE 15 now (as\non master). This allows to remove the nodes from CI.\n\nDepends-On: https://review.opendev.org/677947\nChange-Id: Ic5083ab9b2c14e528f8a41415279ff74a3511313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4e4edc2b9f2c5f38066381a1f486100795b5a1b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4e4edc2b9f2c5f38066381a1f486100795b5a1b8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
