)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~683790","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fqueens~Ic5083ab9b2c14e528f8a41415279ff74a3511313","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/queens","topic":"goodbye_42.3","hashtags":[],"change_id":"Ic5083ab9b2c14e528f8a41415279ff74a3511313","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-09-21 14:52:57.000000000","updated":"2019-10-08 12:48:18.000000000","submitted":"2019-10-08 12:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683790-1570538898832-fe29816b","meta_rev_id":"bb342eedf7f909e4ae1f852a1efadefb4729266d","_number":683790,"virtual_id_number":683790,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 12:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-10-08 12:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-26 20:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"date":"2019-09-21 14:52:57.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-10-08 12:30:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-21 14:52:57.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-26 20:30:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-08 12:30:26.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-08 12:48:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db9ad15f8d3d1b0c3908eff4c53c8de8d1744f7e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-21 14:52:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd21938c3e014266397604cac71065ea2447ddba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 15:21:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2807259dc1ba4f21a6dd0910669b19fa : SUCCESS in 3m 39s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/35fac1f9a86c4eb2858a3023203ae84e : SUCCESS in 8m 31s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e739c091e0bd403caccc32d56dd07c6b : SUCCESS in 10m 57s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/cb8035cbec5544e39205be56516a16b1 : SUCCESS in 11m 02s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/b19e2b01fed1450d83f4569d66d04afd : SUCCESS in 22m 09s\n- openstack-ansible-lxc-dir-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b35b31abb4ad4b929ee3923e5f99b705 : SUCCESS in 10m 20s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a46f40f77dee48af9a8d467d2ae89709 : SUCCESS in 11m 48s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b15116eb4356453f9d7e5a2479069321 : SUCCESS in 10m 09s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/43c2888ec59142bea7500f37d6051e6b : SUCCESS in 16m 56s\n- openstack-ansible-lxc-zfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/58843389899146e4ab904f0428eb6b2a : SUCCESS in 17m 08s","accounts_in_message":[],"_revision_number":1},{"id":"dc7e315aef32ad4d19e54ad301c20e9ba150a61e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-26 20:30:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"628641a4c4ef820c48a03ea241e355ae5fdcaeed","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-08 12:30:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a077fd4631644829489617c617d8441d8a983662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:30:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c01426acc8c967015f4d05ed18787fa55eb5f005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:48:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bbc20cdf1c4644e4888f0632681ff7bb : SUCCESS in 3m 37s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8559f9cdd3c446ba9bda8adc4755f671 : SUCCESS in 5m 52s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/cf2df51d4ea84b239ee4aa490bb1bb02 : SUCCESS in 9m 59s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d5fdfcd692194ebe8b936b95f07bcb41 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"eda1181ac63f057b46dfd74930fede1821762783","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:48:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d7413e7f0369f3e82447ea80ad2cda72129713e","revisions":{"3d7413e7f0369f3e82447ea80ad2cda72129713e":{"kind":"REWORK","_number":1,"created":"2019-09-21 14:52:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/90/683790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/90/683790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/683790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/683790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/683790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/683790/1"}}},"commit":{"parents":[{"commit":"c8095fc486252cae2b401e9ca2fe23b5f7d72dde","subject":"Dynamically use lxc 2/3 syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c8095fc486252cae2b401e9ca2fe23b5f7d72dde"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:53:18.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-21 14:52:40.000000000","tz":120},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job, we test with openSUSE 15 now (as\non master). This allows to remove the nodes from CI.\n\nChange-Id: Ic5083ab9b2c14e528f8a41415279ff74a3511313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d7413e7f0369f3e82447ea80ad2cda72129713e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d7413e7f0369f3e82447ea80ad2cda72129713e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
