)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~683791","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fpike~Ic5083ab9b2c14e528f8a41415279ff74a3511313","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/pike","topic":"goodbye_42.3","hashtags":[],"change_id":"Ic5083ab9b2c14e528f8a41415279ff74a3511313","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-09-21 14:54:50.000000000","updated":"2019-10-14 16:30:34.000000000","submitted":"2019-10-14 16:30:34.000000000","submitter":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683791-1571070634012-f62c7b6a","meta_rev_id":"b20e2e1d4a8f75e4e061f28e5ccea2ddf55faab0","_number":683791,"virtual_id_number":683791,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-10-14 16:30:22.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":-1,"date":"2019-09-21 15:24:33.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-10-14 16:30:22.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-10-14 16:30:22.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"date":"2019-09-21 14:54:50.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-21 14:54:50.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-21 15:24:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-14 16:30:22.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"098ea6278847d80d6c9cb8ff5b2e8b777d5c9a81","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-21 14:54:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c7513cd222827922c7496bc6d72db9739102d49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 15:24:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c65857c8a57344e2ad11a896448b95c9 : SUCCESS in 4m 22s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3a08688ae9f547ccad5173243c901750 : SUCCESS in 7m 53s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/707b2918c1d344529e239ce16bcda173 : FAILURE in 10m 20s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1761cb8a5dde43d993a851e171aae884 : SUCCESS in 17m 17s\n- openstack-ansible-lxc-dir https://zuul.opendev.org/t/openstack/build/f91cae79142546a3a10c9117d716ca05 : SUCCESS in 20m 14s (non-voting)\n- openstack-ansible-lxc-btrfs https://zuul.opendev.org/t/openstack/build/c170e763c3254edc8891bbe91967da55 : SUCCESS in 25m 56s (non-voting)\n- openstack-ansible-lxc-machinectl https://zuul.opendev.org/t/openstack/build/f7d3899b8bfe428b942b8f0e71f721d3 : SUCCESS in 11m 09s (non-voting)\n- openstack-ansible-lxc-overlayfs https://zuul.opendev.org/t/openstack/build/68e6429a21384493985e1c759cbfaec6 : SUCCESS in 25m 42s (non-voting)\n- openstack-ansible-lxc-zfs https://zuul.opendev.org/t/openstack/build/1d3678b51b8b47c8a5b8f648cfe49672 : SUCCESS in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d91313443a6ba68ca3ec1ee5562687fcd7461cad","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-14 16:30:22.000000000","message":"Patch Set 1: Verified+2 Code-Review+2 Workflow+1\n\nThe opensuse-423 nodeset is being removed.  This change is being force-merged to correct the resulting configuration error.","accounts_in_message":[],"_revision_number":1},{"id":"d2befd7a29cbf5befd90a55be464fda3b5dacc39","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-14 16:30:34.000000000","message":"Change has been successfully merged by James E. Blair","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eda0e0d8c035932568cb70619712aa2f6237647c","revisions":{"eda0e0d8c035932568cb70619712aa2f6237647c":{"kind":"REWORK","_number":1,"created":"2019-09-21 14:54:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/91/683791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/91/683791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/683791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/683791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/683791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/683791/1"}}},"commit":{"parents":[{"commit":"12793c336ba7680fe35f93c56552942761fb50d9","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/12793c336ba7680fe35f93c56552942761fb50d9"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:53:18.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-21 14:54:10.000000000","tz":120},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job, we test with openSUSE 15 now (as\non master). This allows to remove the nodes from CI.\n\nChange-Id: Ic5083ab9b2c14e528f8a41415279ff74a3511313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eda0e0d8c035932568cb70619712aa2f6237647c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eda0e0d8c035932568cb70619712aa2f6237647c"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
