)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~684869","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Iff688aa9c3f56d901d95e25c5139dc34cc3253d5","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"constraints","hashtags":[],"change_id":"Iff688aa9c3f56d901d95e25c5139dc34cc3253d5","subject":"Update the constraints url","status":"MERGED","created":"2019-09-26 02:10:21.000000000","updated":"2019-09-26 21:28:44.000000000","submitted":"2019-09-26 21:27:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684869-1569533239721-9cafdce8","meta_rev_id":"7e9b774aae812decfbddc2425437774e41a4acb1","_number":684869,"virtual_id_number":684869,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 21:27:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-26 14:08:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-26 20:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-26 20:56:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 14:08:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-26 20:56:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-26 21:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f0e2bb88811c9307d63bf11d3afc93171ae1fb3","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 02:10:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8227c648acff6a560d54f54da23c3e5ebeb05f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:05:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/62f41f21822445b382241ae9ec3f48a8 : SUCCESS in 7m 06s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/24692a675222474388bbdb61e7b69614 : SUCCESS in 11m 43s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e55609d212314266861ae9458c7549c7 : SUCCESS in 11m 59s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c4deaf8c8e6427ea572dc45ca39cd71 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e54b89d58f42ca9be0673e46bb7588 : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7906efacd301464f87609660401eddeb : SUCCESS in 5m 01s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/5bbeaf5dcca041bda78aef6f0c85105e : SUCCESS in 18m 34s\n- openstack-ansible-lxc-dir-opensuse-150 https://zuul.opendev.org/t/openstack/build/95866b96dad641d895469e68935deb97 : FAILURE in 15m 43s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d568027c46de4775adb66f6d110ef01b : SUCCESS in 15m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef370fe747904e80b5832b4f99da0250 : SUCCESS in 18m 39s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/cfce223ba40e42d5b80c425a020ac01c : FAILURE in 10m 25s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a273eb28f318476d87f3fd8667aaa453 : SUCCESS in 13m 33s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/7fe94d884e3f4e229a64859f8a91b01c : SUCCESS in 17m 59s\n- openstack-ansible-lxc-machinectl-opensuse-150 https://zuul.opendev.org/t/openstack/build/bfde3d43f30b4778b4ade96f8d560648 : FAILURE in 10m 27s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0de4e6928a514549a69f63a4f0cd72a8 : SUCCESS in 14m 45s\n- openstack-ansible-lxc-overlayfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/13bc83b74b0d4a048d2542d377dff945 : FAILURE in 12m 14s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d861d7aaaa7403697c38f9d4cf30702 : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ae96e37ef3f78b46a09527796aeefc895f7730e0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-26 14:08:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acb4b1b90d144c4481d8817e9a1ecc7ff0600410","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-26 20:56:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c941a604cda5b06ec818f9c7166e94bbb369cb5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 20:56:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9d2e737f5569559a57efc3e72f4cacf4e601797","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 21:27:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9f0efeb3133b2cc234911217f60fbaae48f5a87d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 21:27:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cfadec111ca34b399a1bd83ce2713a73 : SUCCESS in 5m 17s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/119ef99114f2498da85ad59f6d2eca10 : SUCCESS in 7m 48s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/204555080abc45499b622bcf5d960231 : SUCCESS in 10m 33s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1391e0e8fa5a4c7eb294414930db2a18 : SUCCESS in 17m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a5b87c53814e1bb83c99f3bd6c3957 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/283cd8d6f83c486ca20b3d6fdab165a7 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d20b28966d008a0b6b4d0fc04cc9d3b217b8c0e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 21:28:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4008b062cc41e2986fbda9072b2dc3 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6017470bd0f4f05a4b35e4569cf5430 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03089fa08e40fc3c8f93f95df7bfd3742d9cbc56","revisions":{"03089fa08e40fc3c8f93f95df7bfd3742d9cbc56":{"kind":"REWORK","_number":1,"created":"2019-09-26 02:10:21.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/69/684869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/69/684869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/684869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/684869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/684869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/684869/1"}}},"commit":{"parents":[{"commit":"3d194c6de30411ed65950bb4c5bbde3ccd42b85c","subject":"Use version test instead of version_compare","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d194c6de30411ed65950bb4c5bbde3ccd42b85c"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 02:10:23.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 02:11:05.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Iff688aa9c3f56d901d95e25c5139dc34cc3253d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/03089fa08e40fc3c8f93f95df7bfd3742d9cbc56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/03089fa08e40fc3c8f93f95df7bfd3742d9cbc56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
