)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~689012","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I1a8b7ec942eb594c669c3feb0f2b6db159d8b909","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"opensuse-150","hashtags":[],"change_id":"I1a8b7ec942eb594c669c3feb0f2b6db159d8b909","subject":"Remove openSUSE 15 jobs","status":"MERGED","created":"2019-10-16 19:21:27.000000000","updated":"2019-10-29 18:15:21.000000000","submitted":"2019-10-29 18:12:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689012-1572372728441-38f9f3a5","meta_rev_id":"2a741cfb943eb7b277e9424891f2aceb0ac5e05b","_number":689012,"virtual_id_number":689012,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 18:12:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-10-18 09:14:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-10-29 17:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 17:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"date":"2019-10-29 17:15:12.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-10-29 17:01:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 09:14:04.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-10-18 17:09:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-29 17:01:20.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-29 18:12:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53d8c0c8e901946ef42a44824a521e52ac9cc0ed","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-16 19:21:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6b2bae39290e6b59ddafe02a26e77cb5a630e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 20:43:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/39d275bd92d94498bc224548d925fca7 : SUCCESS in 5m 40s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/7d354769394747a3910e5a4ae1a68d9c : SUCCESS in 10m 44s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c9968c10605f4812a56c15325d8dfdbc : SUCCESS in 12m 09s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe8df945fa6e44d59d5f23ad92ca021b : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a26408efa247a894a7d7d62b4c3db6 : SUCCESS in 3m 28s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/5f4ba829e8ac4f348b1fb34d12193792 : SUCCESS in 17m 58s\n- openstack-ansible-lxc-dir-opensuse-15 https://zuul.opendev.org/t/openstack/build/92a5782804ec444191946509aedcafa0 : FAILURE in 13m 09s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef78c2c27531436081e0b42dfb3adfc6 : SUCCESS in 15m 58s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a41c05c9395f48258b2644a422d27336 : SUCCESS in 11m 42s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-15 https://zuul.opendev.org/t/openstack/build/c700644b15254f52b0a0d293bad9f7da : FAILURE in 13m 34s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0209fd9a69154ab0a5679ec610000a3f : SUCCESS in 9m 12s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/e2b93f34da034b0cab7bf45cc698b58c : SUCCESS in 20m 42s\n- openstack-ansible-lxc-machinectl-opensuse-15 https://zuul.opendev.org/t/openstack/build/c344096ec80a4cf6bf9d18c8f51fb5ef : FAILURE in 15m 18s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9eb975c96fc4258a8d53c9eddb9a994 : SUCCESS in 13m 35s\n- openstack-ansible-lxc-overlayfs-opensuse-15 https://zuul.opendev.org/t/openstack/build/200bb480ee1648cea7e545aa6d045d9a : FAILURE in 11m 51s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa2e5cec49844578816f01c96deef266 : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d49d9e7c2f5476777d2ec4778b7188f224fcb7d6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-18 08:34:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4795fad9a1b2e04295c54f0bc44a93caa6efecae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 09:07:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aff2feebda7f403c8ba0f87f2f6a90f2 : SUCCESS in 6m 50s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/eae0d4bc6e214f91b7ae4a0bc2033835 : SUCCESS in 7m 47s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6e499d654830405dad2c562b24d1eb7c : SUCCESS in 9m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7756734d2a044e248a9a6c959cc223f1 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4135adbb3b704fd0a545ec4715b771f8 : SUCCESS in 5m 04s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/1343722fa92e4d1ba63f3fe7b62f1357 : SUCCESS in 15m 30s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64bf0c1c6be04ce2833bd033ae0cc193 : SUCCESS in 17m 57s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35712a06bfba446eb3d26932f8fff8a5 : SUCCESS in 12m 21s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbd6ec7d11114dbb87b6804af0f3bbb6 : SUCCESS in 11m 29s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/c3d7279082014ce79596d6318f78cf37 : SUCCESS in 15m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cdf4283fecf42afa419729664b0816a : SUCCESS in 13m 38s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3d56ab94eb04d5483a682d72ad1cbe4 : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":2},{"id":"681167c918e228bf22457e4afbd167c5f7a77e41","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-18 09:14:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b7451fb6ce09b85c795e9a0fa347038eee4823e2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-18 17:09:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dbafa80c0e584bdd02857d2bd6ca29c1573bb288","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-29 17:01:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a931a2602ce1283299fb3d920c003fdde6a7683d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 17:01:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"028aa1b842a912464e2fea42185b425ce3cea1a7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-29 17:14:32.000000000","message":"Patch Set 2: Workflow-1\n\nSorry, need to double check something first","accounts_in_message":[],"_revision_number":2},{"id":"9ab73fa177caa946cd5afd6fefd8ddd81d0f1bdf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-29 17:15:12.000000000","message":"Patch Set 2: -Workflow\n\nto be clear: the lxc version is not supported","accounts_in_message":[],"_revision_number":2},{"id":"631fcde125e1ada621f2a9ae0be2e79711a57455","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:12:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f4be0f0a917232ccb6ec6814a1aba7c8ecd47bbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:12:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f5662433fcee4259a54d06d8b1e61624 : SUCCESS in 6m 56s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/ab72d7b6d5cf4454a6d9b9d6c86ddeeb : SUCCESS in 11m 36s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/bc69d2ef6d424019b9bf86ee82b32676 : SUCCESS in 17m 34s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc8beb18f0a24c2abb9a0b1f0e5f1264 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43abeccc1b5f428397439d75e9771c77 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f4b3e18f60aea0bc6bb5f2c6569b03c4572db6f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:15:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/421f7c26fe874564bb235dbf553dc304 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7cdb2d1ac34f20b6369252c98c181f66b40d7667","revisions":{"b6c74a6b8a4a87a8e272b0558b11a3e8adb4e92b":{"kind":"REWORK","_number":1,"created":"2019-10-16 19:21:27.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/689012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/12/689012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/689012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/689012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/689012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/689012/1"}}},"commit":{"parents":[{"commit":"62c3a2cbe177cd32072bce92116b58ac247c2f6d","subject":"Add Centos GPG key custom url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/62c3a2cbe177cd32072bce92116b58ac247c2f6d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:21:17.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:21:17.000000000","tz":120},"subject":"Switch to opensuse-15 nodeset","message":"Switch to opensuse-15 nodeset\n\nopensuse-150 nodeset is referring to openSUSE 15.0, which is still in\nmaintenance but openSUSE 15.1 has been released already. \"opensuse-15\"\nis going to refer to the \"latest openSUSE 15.x\" build released and\nworking for OpenStack going forward, so add this nodeset and use\nit by default going forward.\n\nChange-Id: I1a8b7ec942eb594c669c3feb0f2b6db159d8b909\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6c74a6b8a4a87a8e272b0558b11a3e8adb4e92b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6c74a6b8a4a87a8e272b0558b11a3e8adb4e92b"}]},"branch":"refs/heads/master"},"7cdb2d1ac34f20b6369252c98c181f66b40d7667":{"kind":"REWORK","_number":2,"created":"2019-10-18 08:34:48.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/689012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/12/689012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/689012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/689012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/689012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/689012/2"}}},"commit":{"parents":[{"commit":"62c3a2cbe177cd32072bce92116b58ac247c2f6d","subject":"Add Centos GPG key custom url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/62c3a2cbe177cd32072bce92116b58ac247c2f6d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:21:17.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-18 08:34:41.000000000","tz":120},"subject":"Remove openSUSE 15 jobs","message":"Remove openSUSE 15 jobs\n\nopenSUSE does not support lxc, remove all jobs.\nThere\u0027s no value testing a use case that will not work.\n\nChange-Id: I1a8b7ec942eb594c669c3feb0f2b6db159d8b909\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7cdb2d1ac34f20b6369252c98c181f66b40d7667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7cdb2d1ac34f20b6369252c98c181f66b40d7667"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
