)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~689103","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fstein~Ia30f20df6971a9a44a69e5cc22020831a95a1489","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/stein","topic":"centos-private-epel-stable/stein","hashtags":[],"change_id":"Ia30f20df6971a9a44a69e5cc22020831a95a1489","subject":"Add Centos GPG key custom url","status":"MERGED","created":"2019-10-17 09:02:44.000000000","updated":"2019-10-24 21:21:04.000000000","submitted":"2019-10-24 21:20:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689103-1571952002991-fee2ba80","meta_rev_id":"f366542d2a3d1c1d4064104b82e0542af26d7840","_number":689103,"virtual_id_number":689103,"owner":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 21:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-23 11:36:22.000000000","_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},{"value":2,"date":"2019-10-22 09:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":2,"date":"2019-10-24 21:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"date":"2019-10-23 11:34:34.000000000","_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":1,"date":"2019-10-24 21:03:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 09:00:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-23 11:34:34.000000000","updated_by":{"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},"reviewer":{"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},"state":"REVIEWER"},{"updated":"2019-10-24 21:03:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-24 21:20:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61e8827c28e5a41ce6218ca4ca87367b1e904e81","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 09:02:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fe7b6918a64d7968de5ccb745c48fbeabc4bd3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 09:23:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c7dfc72d35854ae598053d469d722cf7 : SUCCESS in 5m 45s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/0590fc23899e40b098b2db9aa9eb6af4 : SUCCESS in 8m 46s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8af6e6e7d7654a3091ea5baf93e9fe2d : SUCCESS in 9m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a20f5af2f34a4880b2c241323d6b3122 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b80e41befbd4dd99913260a3bc86a8e : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a803554bfca94eac90c9b0bf4c3a5bfe : SUCCESS in 3m 29s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/d54537f7fecb4f169da9d9fbe5f65e16 : SUCCESS in 14m 58s\n- openstack-ansible-lxc-dir-opensuse-150 https://zuul.opendev.org/t/openstack/build/8307bfd127df47cea9cf85127b0c0c49 : FAILURE in 13m 58s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f6083ed8d15460c909c8415bc73ac52 : SUCCESS in 11m 02s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63ed46c11983451fac0709a5482cccc1 : SUCCESS in 12m 05s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/83282249b40f405f9ea28f964ff319ee : FAILURE in 15m 10s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a8f17f1695a455bb193088bcad31ac4 : SUCCESS in 11m 37s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/b5e6e1a2aa594f09bbfd1b0e37988a66 : SUCCESS in 14m 54s\n- openstack-ansible-lxc-machinectl-opensuse-150 https://zuul.opendev.org/t/openstack/build/48208fc8a7d9431aa26fafe3bc52fc6c : FAILURE in 12m 32s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c92de82afd824e089c28b74c4e19de4e : SUCCESS in 13m 46s\n- openstack-ansible-lxc-overlayfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/a1ab501d00194f4d9ad931667b362c5b : FAILURE in 15m 00s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/425e9dd16b444b33a472438f9a3df665 : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e5a3f3af08fa1ccbc6daacfe00318e84f331bb5b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 10:51:28.000000000","message":"Patch Set 1: Code-Review+1\n\nWe can backport it, but we will need to backport https://review.opendev.org/689122 as well then","accounts_in_message":[],"_revision_number":1},{"id":"056b77a518ce628acd2a842771b9a1d3fa17b32e","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 14:57:22.000000000","message":"Patch Set 1: Code-Review-1\n\nThis change should add https://review.opendev.org/689191 what should I do ?","accounts_in_message":[],"_revision_number":1},{"id":"1219f98fd63581e4ddd7dfa2a8362cf1bf3f5a8b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-21 13:41:18.000000000","message":"Patch Set 1:\n\nSo we need to merge this cherry-pick and after it we should do another one from https://review.opendev.org/#/c/689191\n\nOr, you can edit this patch to implement changes from 689191","accounts_in_message":[],"_revision_number":1},{"id":"728bcbe8e0f7ac4df0309132799f6d0cb478b1d1","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-21 13:53:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d88926ad30a507ea8c484120805e05a2d3f57c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:27:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8f13b2fdce914e39a0e07d7dbfe0e179 : SUCCESS in 5m 24s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/5ee314c901884f2a9b0a344374a26dd9 : SUCCESS in 8m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/16d4c24745004707ab259d52ab36857d : SUCCESS in 16m 12s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f17b49d6fd64015a75b41b89ae3e258 : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f515298a3d49ae964ddc2fef9f1cb7 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71774052e16c4cb8882b0d080e530184 : SUCCESS in 4m 48s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/04d473683036408eba6a8a4404ce10f0 : SUCCESS in 15m 14s\n- openstack-ansible-lxc-dir-opensuse-150 https://zuul.opendev.org/t/openstack/build/3fcdd891228443f2ad7d3d0e4f31a5c3 : FAILURE in 14m 08s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17aba6dfcf5841b0adfd2473b2dbaa41 : SUCCESS in 14m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a29912a7127c4f93a2d873f430873173 : SUCCESS in 11m 06s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/63e9a7410d664f5fba0a89327fe0dc26 : FAILURE in 16m 23s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d0d030fdf9f418ebc3a87acdc70ab50 : SUCCESS in 11m 36s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/acfb52e1b9fa4a6f9842fbf7cf2a0eab : SUCCESS in 13m 27s\n- openstack-ansible-lxc-machinectl-opensuse-150 https://zuul.opendev.org/t/openstack/build/8295904a80d24e0ea232d9984581f427 : FAILURE in 10m 54s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b746dfa3a1954c089be9c219086490ad : SUCCESS in 12m 12s\n- openstack-ansible-lxc-overlayfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/c5bc17241bf841368cd0ef8e11b8c3e5 : FAILURE in 12m 43s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfbe4a21586b498f92da0d70fa6ea9dd : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ee97a025fe1e7be0f2e47df2709ccde475cd830b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-22 09:00:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f04268927199e7aee0e15f27cb0fa2bb0c1cbb14","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-23 11:36:22.000000000","message":"Patch Set 2:\n\nHi Jonathan, can you check this, that\u0027s the two changes merged on master for epel key.","accounts_in_message":[],"_revision_number":2},{"id":"237b2cce2e57e3371e43c19711ec215938188d6e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-23 11:42:08.000000000","message":"Patch Set 2:\n\nyou\u0027d better ping in IRC:)","accounts_in_message":[],"_revision_number":2},{"id":"e6d104310c4c37d94fb1c759f6950d16bc036229","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-24 21:03:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c6e0b8e899f4be8503d157b0a7b251b5d5cd2834","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:03:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7614015e0c0c5025788d7bdfe0ea6268af9be5e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:20:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"76895b9bdcb92b793c711cf4a8778e4e7321eaa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:20:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f0c0da865e1d4bbf880f9e8a37456441 : SUCCESS in 6m 48s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/df3ac3e2fabf4285956e72fd62ccccd1 : SUCCESS in 7m 40s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e92c6b3a39cf4e07b6637e5280277159 : SUCCESS in 11m 05s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/264983887054409e866dd2c515f91681 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db0457dae1747efaedda6e466c80816 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2804393aff44b44a4d6d987850b2a40 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e8e1d9df092cf9ed9b8c121e31e953e46cb426d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:21:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a426be820d496bb8e3160a685535e0 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f1c38c46e3f493685fc56daacb47538 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cce5c54363593615447c83ffbc90645e0aa48307","revisions":{"26dc54bfed8628ea2d9d045631c3648f5021968d":{"kind":"REWORK","_number":1,"created":"2019-10-17 09:02:44.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/03/689103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/03/689103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/689103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/689103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/689103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/689103/1"}}},"commit":{"parents":[{"commit":"32ede70043039fca1e5ac2e966d07d827e43a14f","subject":"Fix copying apt keys from host to containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/32ede70043039fca1e5ac2e966d07d827e43a14f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 16:16:44.000000000","tz":120},"committer":{"name":"kourosh vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 09:02:44.000000000","tz":0},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: lxc_centos_epel_gpg_key\n\nChange-Id: Ia30f20df6971a9a44a69e5cc22020831a95a1489\n(cherry picked from commit 62c3a2cbe177cd32072bce92116b58ac247c2f6d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/26dc54bfed8628ea2d9d045631c3648f5021968d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/26dc54bfed8628ea2d9d045631c3648f5021968d"}]},"branch":"refs/heads/stable/stein"},"cce5c54363593615447c83ffbc90645e0aa48307":{"kind":"REWORK","_number":2,"created":"2019-10-21 13:53:07.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/03/689103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/03/689103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/689103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/689103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/689103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/689103/2"}}},"commit":{"parents":[{"commit":"32ede70043039fca1e5ac2e966d07d827e43a14f","subject":"Fix copying apt keys from host to containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/32ede70043039fca1e5ac2e966d07d827e43a14f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 16:16:44.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-21 13:53:05.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: lxc_centos_epel_key\n\nChange-Id: Ia30f20df6971a9a44a69e5cc22020831a95a1489\n(cherry picked from commit 62c3a2cbe177cd32072bce92116b58ac247c2f6d)\n(cherry picked from commit 921c98f67bb6cf61fbb7d1393245ab9e4aac8bdb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cce5c54363593615447c83ffbc90645e0aa48307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cce5c54363593615447c83ffbc90645e0aa48307"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
