)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~689191","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Icf84a0a55654fa890947bae5b608870eddad7324","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"centos-private-epel","hashtags":[],"change_id":"Icf84a0a55654fa890947bae5b608870eddad7324","subject":"Centos EPEL options default to global variable","status":"MERGED","created":"2019-10-17 14:55:12.000000000","updated":"2019-10-17 18:13:50.000000000","submitted":"2019-10-17 18:12:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689191-1571335962296-ed58b23b","meta_rev_id":"33cda81f5cfc7c0cfb553a92f6d7083be2e829b4","_number":689191,"virtual_id_number":689191,"owner":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 18:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2019-10-17 15:27:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-17 17:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"date":"2019-10-17 15:32:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-17 17:59:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 15:27:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-17 15:32:20.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-10-17 17:59:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-17 18:12:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a9cdfb21c78d9530d6e0f55ffe170f26c483ebf","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 14:55:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f9552cf6b76a4318017faf0ad8f86f5f3b57054","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 14:56:12.000000000","message":"Patch Set 1:\n\nIt\u0027s dublicated patch https://review.opendev.org/689122\nSo you may abandon it.","accounts_in_message":[],"_revision_number":1},{"id":"255d18a09ce37d3c2930230e77c3c01c1bcc773c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 14:57:42.000000000","message":"Patch Set 1:\n\nbtw reno is ok as lxc_centos_epel_mirror is still valid, and centos_epel_mirror _may_ be used to override everywhere","accounts_in_message":[],"_revision_number":1},{"id":"094e65d4df6b015ebcdf4a204b7e936e847f95a5","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 15:05:48.000000000","message":"Abandoned\n\nduplicate https://review.opendev.org/689122","accounts_in_message":[],"_revision_number":1},{"id":"d3abb9cecaa2856f7051890a33397229e2e7529f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 15:25:25.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"52e6e8c33d71cdc39ddf07a49d8cacfe55387972","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 15:26:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7c5275148ba95d9c5a0d059ce4abef96a0a26cfd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 15:27:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6dfb9684a546f284c15c1da91e3334bdee4df1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 15:51:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f0e46f2364a94a50a64364043c9ae706 : SUCCESS in 6m 01s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/af1124fe156d49f7b99ab41232e2c738 : SUCCESS in 7m 25s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9cfb626a60f14183b68004a373124ddf : SUCCESS in 10m 04s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42ddfbf0316942c982450a4a29bfa5aa : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a21cb1ab98471d907c40b80727ab8b : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97dc0777f061454d88bed955f945f500 : SUCCESS in 3m 46s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/57a2cca96a224500b8dbf37bb6600c47 : SUCCESS in 17m 39s\n- openstack-ansible-lxc-dir-opensuse-150 https://zuul.opendev.org/t/openstack/build/dd85f2e0b49f44da99042cd99a557bce : FAILURE in 15m 51s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c75c9793d6c64b609ee2ddfbebd89245 : SUCCESS in 12m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61aa81edac014e4986fffa04d53d31d6 : SUCCESS in 11m 27s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/8695bc181e16472c9a262f5e41672b57 : FAILURE in 17m 45s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6adb3e42ecec414eab5640bb2605c3d0 : SUCCESS in 12m 26s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/ee6866e6a78240e58635e1e3afb47c92 : SUCCESS in 14m 33s\n- openstack-ansible-lxc-machinectl-opensuse-150 https://zuul.opendev.org/t/openstack/build/a4afa98d296543feb29aa633ad95f086 : FAILURE in 12m 58s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66f14ba4089440829a1cbdfc34cd719a : SUCCESS in 12m 24s\n- openstack-ansible-lxc-overlayfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/5d60b58c2e134595a8d56a983dd4b09a : FAILURE in 18m 46s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86078198296b4400a8c9b0174ece40a5 : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1b35ce3c8c837a9e72a15f13316030f09edcd093","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-17 17:59:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c291ebe16e95f1948cff6ad6d37abd5d5933e262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 18:00:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"406289a4bfd6fe1f816cc9fa7bcaa77450460f8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 18:12:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c07c6bb8b0b04f7f9aecff5d69a8485a : SUCCESS in 6m 54s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/aad64bda15fe46b4a68d179233aa3e8e : SUCCESS in 8m 53s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a01dd29e8edc4ca49c4881046817ca72 : SUCCESS in 11m 11s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c71bc2685814d1c8a70e78de353e049 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a2fac08eac41f8b99744a00a545c61 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/559efd808ea645d1be0c9a8f8ee7304e : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4f8112b75d2d8f1f30177806affa8cea19747737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 18:12:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1982b0fc4dc524a356342df4d01912d6190509b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 18:13:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c74a4a768a74ea2a207cf0818ada0ca : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/707e08c638dc44bc881a9d7d892bf232 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"921c98f67bb6cf61fbb7d1393245ab9e4aac8bdb","revisions":{"71dbb7d3b375723fa9db81ae5c183ebe3dd0fb51":{"kind":"REWORK","_number":1,"created":"2019-10-17 14:55:12.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/91/689191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/91/689191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/689191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/689191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/689191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/689191/1"}}},"commit":{"parents":[{"commit":"62c3a2cbe177cd32072bce92116b58ac247c2f6d","subject":"Add Centos GPG key custom url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/62c3a2cbe177cd32072bce92116b58ac247c2f6d"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 14:48:16.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 14:54:33.000000000","tz":120},"subject":"Centos EPEL options default to global variable","message":"Centos EPEL options default to global variable\n\nRepository variables lxc_centos_epel_mirror and lxc_centos_epel_key will\ndefault to centos_epel_mirror and centos_epel_key\n\nChange-Id: Icf84a0a55654fa890947bae5b608870eddad7324\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/71dbb7d3b375723fa9db81ae5c183ebe3dd0fb51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/71dbb7d3b375723fa9db81ae5c183ebe3dd0fb51"}]},"branch":"refs/heads/master"},"921c98f67bb6cf61fbb7d1393245ab9e4aac8bdb":{"kind":"REWORK","_number":2,"created":"2019-10-17 15:26:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/689191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/91/689191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/689191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/689191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/689191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/91/689191/2"}}},"commit":{"parents":[{"commit":"62c3a2cbe177cd32072bce92116b58ac247c2f6d","subject":"Add Centos GPG key custom url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/62c3a2cbe177cd32072bce92116b58ac247c2f6d"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 14:48:16.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-10-17 15:26:12.000000000","tz":0},"subject":"Centos EPEL options default to global variable","message":"Centos EPEL options default to global variable\n\nRepository variables lxc_centos_epel_mirror and lxc_centos_epel_key will\ndefault to centos_epel_mirror and centos_epel_key\n\nChange-Id: Icf84a0a55654fa890947bae5b608870eddad7324\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/921c98f67bb6cf61fbb7d1393245ab9e4aac8bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/921c98f67bb6cf61fbb7d1393245ab9e4aac8bdb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
