)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~690342","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Ftrain~Id051ad4936b4ed8f6a5c7eb3860ba793dc7fbbcd","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/train","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Id051ad4936b4ed8f6a5c7eb3860ba793dc7fbbcd","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-10-22 19:14:35.000000000","updated":"2020-10-09 13:49:55.000000000","submitted":"2020-10-09 13:48:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690342-1602251297849-c901a1f2","meta_rev_id":"6853782db432244c7fd43aabac4c93d373752e80","_number":690342,"virtual_id_number":690342,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 13:48:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-10-08 16:18:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-10-09 13:28:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 09:18:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-09 13:28:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-09 13:48:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8bf7bda87aae3b25babecad2a699774034bac73","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-22 19:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b09849e6d1166e88063d0b1a46c22b0b82399cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:30:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8b152b75cf2d4a459d49000f1379683a : SUCCESS in 7m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/711b84bccda34476a77ee92853863086 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c3ab829e134fb8a2c574ab0d4aeb2b : FAILURE in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a51045a1baa54d5d861ffd1e89dddcce : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e76b1907d7a25c1d30883bda8812139bf9612021","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-24 18:32:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45e8f58c14c437b58ceeb82f1f67ffb79574d497","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-24 19:00:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a5dc1260a2a0ea02402d095e0d3204cea9fbb4ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 19:17:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8826854735a442febb8a8addfeb46efb : SUCCESS in 12m 05s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/559a2f0f378644f993f7aeed745b8ccb : FAILURE in 4m 30s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2ad2044b626b4f5685e546b03f570cc0 : FAILURE in 4m 54s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee6f887466bd4f498808531e2cfee9d7 : FAILURE in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e8e514a99a4554bece291a4a3d69c9 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/099b9d96b79e4ed48b62a47d4fcbcc3f : SUCCESS in 5m 00s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/e9b5dec43a5146f784ea67b92a50b784 : SUCCESS in 15m 30s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d28aeda3c9d04b408fc7cee8c99dec85 : SUCCESS in 16m 15s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c6e9974f2e64d1292623cc76ab65534 : SUCCESS in 18m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/490c3289bbf84a46b89dad747b330f69 : SUCCESS in 14m 17s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/6c029bbb18b14675b6f34c99125825a3 : SUCCESS in 21m 03s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/541ad62b28ca4c3ea470fabdae5ad027 : SUCCESS in 13m 09s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0472a9801501405b90d4f48c025f55f8 : SUCCESS in 16m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7c387d3a738809911787def11f546cdd8145835d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-02 20:33:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d5c500af4489ecf1276d0e73f8f228528b9ff32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:25:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/261ddfdea90d438c823a5a79975d00fc : SUCCESS in 6m 37s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/174ca23f741c49beb970a510d62221fb : FAILURE in 8m 02s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a80f127446ac4a2b9101c1a2a7334931 : FAILURE in 8m 33s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b5420fc188a4b10960131d9b87e38b7 : FAILURE in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4505dded286495c86f57d1f20875772 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7048b3973d27418d81d79e9cc99d5097 : SUCCESS in 6m 15s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/6eee4097ef684bf095cf0d900437e8fc : SUCCESS in 19m 46s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/931d941f475444239e315be7bab8ef6c : SUCCESS in 12m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc47939b5d914afe82ed3ddb2640f358 : SUCCESS in 22m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1bd9ca366434a21bf38686022624e54 : SUCCESS in 15m 00s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/5f83b7d22ed4416ab017714a07e1c3ff : SUCCESS in 19m 13s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81c2ea6564684f5484dfba2670d86a17 : SUCCESS in 14m 31s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d7d2295d1d44503abd0c2a5d445ac59 : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":2},{"id":"cd9243fb088f0d3e708d3b954b3b8cade2d0ae7b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-23 20:10:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bc4f570ac241e17a4b4071fdef4e1025a1123a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 20:42:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4c5f595cbedd4b29b29fa03e0d02f98b : SUCCESS in 6m 16s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/0ab0abef1c5b4487a57cd5608edf4987 : FAILURE in 9m 00s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/eeaa408b95b4435ab6c44bee52deac2b : FAILURE in 7m 44s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43b878c3ac534e74bbafacedca8f13f0 : FAILURE in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d02a33a74f4d9e8b429cb10902b60e : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c24fd12c0b74f5d9bc7f834c7a01e37 : SUCCESS in 4m 24s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/25c68dbc2fb9403183b59f8e6dee96c0 : SUCCESS in 16m 09s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af2f677dde5e4b9db6a4f1f7489f1fdf : SUCCESS in 18m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d4c27edda2d48b3820d7d757b93fc88 : SUCCESS in 20m 40s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebc036174d4d4667bd1416c27064814b : SUCCESS in 15m 49s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/7adb30b1fe77408aa2462a4638c39ddf : SUCCESS in 14m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0af7cd4cfe14695829cf15f64e5fab7 : SUCCESS in 13m 42s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ce9e1fb13be4ce0a7c7c8628b478e12 : FAILURE in 17m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a763db12d9df9159e393a52174556ee50de19784","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-29 06:29:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2c4d454cac352ccce466702ae66b5c1fca97983f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 06:56:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3ca911ecad914898a575ff1058b6e190 : SUCCESS in 7m 42s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/36bcb661528d482f9341b1fee389381b : FAILURE in 6m 17s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a90d31083ee94945b7fae49ca09fbb6d : FAILURE in 7m 50s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/416859a4d8304430b3935dc29f1c87b1 : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3adc4f8e36dc4440be67165309457fe5 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25198191d6e74f6d8c79d38936e550db : SUCCESS in 5m 31s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/be159d6ed587478f9656378d3b4315b8 : SUCCESS in 18m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3befc531171d4f8aa4570cecf9e5cdd7 : SUCCESS in 13m 30s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9be1cc42d1144d6a9f590eb61bb26f17 : SUCCESS in 16m 19s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dfe199fb78f4e888ed30981dda45c14 : SUCCESS in 17m 35s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/6385c1d68257429fb0ccb8f4841acc48 : SUCCESS in 18m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd991f96709e4da5a44607d5eee6bb37 : SUCCESS in 15m 49s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29f7fb47f5cf4146a81255a037585a8e : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":2},{"id":"73636ec7437f1183602f97e23c97e4d27e8bac66","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-12 19:10:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cad1e8e1e14d6fa970284b90ecf3b9c3d35ea92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:41:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9ac9347d474640b38ee7cf01e6e67422 : SUCCESS in 7m 15s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/d8596c293ed248fcaa767be7a3f10bf8 : FAILURE in 5m 20s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/47c5b9653dc242468409759411d48b1a : FAILURE in 6m 24s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3571990ab8046a48f9eca5c7e7723f1 : FAILURE in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844de49714cf49b7a9254deac7492694 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddc0da0a387045409497f57bfdf2888c : SUCCESS in 6m 23s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/254f73e9b1344d57b978a5399640c0a4 : SUCCESS in 16m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa132e6c25bd4ce9b4f63da102e1e932 : SUCCESS in 13m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4aee3a15f0344f17a340a9e5ee578c4f : SUCCESS in 12m 37s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d275e202bc444d20a54d4288c8980a45 : SUCCESS in 14m 13s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/8b418a37cc0c48cebcd2254ab5966128 : SUCCESS in 17m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e64fb0b3dfd34710be31f806c667f72a : SUCCESS in 20m 33s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1c96eff94be490a959c30498e0f284a : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":2},{"id":"20006b521fc50b1c0c75b8bc43d4b79e2418d50f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-23 13:28:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"84252372d54167a9a581639b5f3aea2f10fafb1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 14:07:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5093a067d3c94b84af8f204f8339d5a6 : SUCCESS in 10m 15s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/43365c044cc44ed5922039a5d23f767b : FAILURE in 11m 14s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c97e93dfeb434363b064ed24ae7b94d5 : FAILURE in 8m 39s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdd517d4f08344b4a7ab5ab4e5c82c4b : POST_FAILURE in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af133895094a485586dbedc5c17b78a7 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66b29e8d24de4e80973b1ac46c719b9e : SUCCESS in 5m 22s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/937d15b43d064fd8b8673d23c9eeeb4b : SUCCESS in 18m 36s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c16493ed2804fd993768d11f969856e : SUCCESS in 25m 10s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79f7bd058d4848e2858cdb67eb2558cc : POST_FAILURE in 19m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90e1d6fe54ca47adba9441ff57a6a00d : SUCCESS in 29m 04s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/41e6b74d41cd4c1a8fc0a99207470f84 : SUCCESS in 36m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2453739194914e2d94dfe36112dd331f : SUCCESS in 30m 27s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acab07083ab44dffb18e8767051a716f : POST_FAILURE in 24m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2765ee24e9c76d73bc030d907bbb3e06a278813a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 10:48:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2943c3650be16c7e3f7c227d7e4fa3940e90cdb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 11:16:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/16313470cb194535ace6cd31c617def7 : TIMED_OUT in 16m 16s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/508636b71d804872b80a5029fc77ae88 : FAILURE in 4m 59s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6a58150fc9c5402e8d83bc7c671b7b4a : FAILURE in 7m 14s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/634acf4d2eb648b4babb1069bdc5f665 : FAILURE in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f321a0fea1d4a3a856afbab700e5032 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcab08c599d04d8291a22271fb10de73 : SUCCESS in 4m 27s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/b9ad7668b0c74487a1cc80d3f337ff8a : SUCCESS in 16m 16s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2e38df570d64e028865e3284a34c695 : SUCCESS in 21m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae575b0ff1124a2ca78fb27786f6d305 : SUCCESS in 13m 16s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf863c213beb4a84b1774ecf604c8eeb : SUCCESS in 23m 15s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/af0431ad186b4ac9aef9f320d4be682c : SUCCESS in 15m 22s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec2dc0f40d7c406494fe32fe25d95bb4 : SUCCESS in 13m 38s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/993fc118945342d8844a1784e6b22d1e : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":2},{"id":"1dde5a50ad481b8ec024e5421262c6fe86fd5fed","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 17:12:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0c41d107b055f2cc455a1b1cde7475c859f65c3b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 17:13:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6b0743e6edd1a492084d1cc3d503f4a2541ad978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 18:02:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1847aec1e06c4db88a708c3c06d2161b : SUCCESS in 7m 17s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/9af5fe381d2c4b8092544ab645a84654 : FAILURE in 5m 13s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1f9f6f04a1a146f990839eb2b5ed727d : FAILURE in 5m 42s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84f5328b1e4c451cab09a23dd0790d38 : FAILURE in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a755544136d497e8f0ce78e01078d54 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89dfaf93d75e436cae4a446f9792b03e : SUCCESS in 4m 51s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/deca6c92d093488eba28feb898f83b64 : SUCCESS in 16m 57s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81a5608c6518472ba6ab00ea634b997b : SUCCESS in 19m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c04f0eef5d6b4743baf033820102c854 : SUCCESS in 18m 51s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03371db9b2124a5eafc969d7b4565ccb : SUCCESS in 17m 56s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/bad764f4872345f9bfd48c923d047672 : SUCCESS in 18m 37s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f11a21d4244945ffb907d827c2343796 : SUCCESS in 16m 18s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce5af8673865451a86cc0b2c070b906b : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a40072b77b92d100d03b3a697dec78953788600b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-27 06:20:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"962827a6b02796016f57a3d538cbb5f0f7b2b408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 06:59:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8e9304c89d6a4be2981d05830ae021c1 : SUCCESS in 8m 09s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/f4293c61e1154c96b4754326c1af7ac1 : FAILURE in 6m 23s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c3fc82df5b0c4582960a315f144670cd : FAILURE in 7m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25c81b40899548248bd5369646579589 : FAILURE in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c6181a94254e39b3594e0c29fb1f19 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/136af4b61cfe4286a1e509ca5e09afd7 : SUCCESS in 5m 14s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/78b8c06bdcfc4f4396f3a7a063fba91a : SUCCESS in 16m 14s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2ddff2c37ca4ed08b0e8cb1604c55d7 : SUCCESS in 12m 14s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/578d2cf7e2a844d083dcd4d2ff85e92c : SUCCESS in 13m 27s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d9d52d90e024b07b9cb861575d7352f : SUCCESS in 16m 08s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/d888385db49246ed80c6dccfcfa5fa55 : SUCCESS in 36m 11s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12d72fc6065a4cce9cc89416e3a0d57b : SUCCESS in 12m 39s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2698fd5b91f44bdca955f7e9cd4ff5f4 : FAILURE in 17m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8b9707ea5c103d5982f9ce8afbf74ffb0ce42af8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-18 05:58:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"26a09564e0c056879b44381ca800fc26a825ee08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 06:39:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d0d15ac68a4143a28954f2b19b0ac55d : SUCCESS in 6m 59s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/440931b61b2c48c88a4a6edf378435b2 : FAILURE in 7m 24s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/88a9fb30a6c94de4a9840a598dbc0c62 : FAILURE in 6m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4661c5ff8d06414b8d08aeba02bf8d55 : FAILURE in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b09d5a12262541d59aa4750134416423 : SUCCESS in 14m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4688576e0eb4d948597c34e081a78ea : SUCCESS in 5m 29s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/e87402c1f5bc4c00938ae833159779ce : SUCCESS in 16m 24s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d85cd5808734efe9872dee3cca96a34 : SUCCESS in 13m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a7ba2ba7d3348a3b118d4b10cc3d838 : SUCCESS in 13m 53s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6b33275afff4d2b8ec5338001981b2e : SUCCESS in 12m 22s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/e9f97dccfc8b4db3b6fbdfe431395261 : SUCCESS in 17m 42s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c9a313347d74a59ad41445b091f5948 : SUCCESS in 13m 03s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db134252b840494faaabb14923b4c3e0 : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":2},{"id":"11b00551f968a89edcfcba39f9d7fc03fdd4bc0a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-26 06:33:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"95a6f0130846b0fa29e3a33dbf463fdf3c43eab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 07:03:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/73567c30266443da91ab81205e085bc6 : SUCCESS in 6m 27s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/23cbcc9615224690bdc349264f8f94ec : FAILURE in 7m 36s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f4d8a59c6951400f8d971d9588742744 : FAILURE in 5m 35s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1fc0cc4738b481093ead159bf367e5f : FAILURE in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276e79141c6a48b7b7f82c970385f68c : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/133756b0e50349c18b03011d205ce27e : SUCCESS in 4m 33s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/d73de6faf6154d978665d231fafbbeeb : SUCCESS in 15m 10s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72d1278c23c14776aff59f3a649ba90f : SUCCESS in 12m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad74f07e5aeb43b496a87e692431e490 : FAILURE in 12m 37s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f4af269bc1e41fe8b40aa6b2c9fbaa3 : SUCCESS in 14m 19s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/661adf274dce450c9a5eb312f13d3c65 : SUCCESS in 16m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1f22229830e49a9a42e50800f81f7a0 : SUCCESS in 14m 11s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f430ca090fe4c359c16e8dc7a58544d : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":2},{"id":"838e4559a85cbbd69c9a06db6ee7d520eda2f066","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-05 19:36:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"63d091bacb06da1f920110b9376f27e2d5f6791d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 20:25:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/386f1711bdd84da7b7a74be804d4de2f : SUCCESS in 5m 52s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/250a6b39566d48d0bbfd7b1017a7d6c3 : FAILURE in 5m 19s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3255567bcd6a496c8314e5346ed17e3c : FAILURE in 6m 05s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76646a159829461d9f870eed0d0322c9 : FAILURE in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efad8d8839c440ddae9ec7c1f3efb012 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a68f3eb12e56484e983113b6bd14971f : SUCCESS in 3m 52s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/e7a4d1acd49e4fba9172c4d49f7225fa : SUCCESS in 16m 47s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7df30f09582b440c8987d91118606919 : SUCCESS in 14m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c71e290896b24a47aa90b524fb5cb680 : SUCCESS in 14m 28s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a05de7f4f75442d9efc12e31c198777 : SUCCESS in 14m 49s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/4afb966c271e464db37997f2ba6c10fe : SUCCESS in 14m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e7ca3439e084731a6a47ed919a290f1 : SUCCESS in 12m 55s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f01cd8fed2341b09c60fcb1488cc012 : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":2},{"id":"96471af5647dc9d0aa076aa43e05d7506555b956","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-06 06:29:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6cc965fca9ef43f8d5ee98ea9dd1ac7357fd806b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:25:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/38075a4c014644f1a27cd9ddbb69f01e : SUCCESS in 6m 02s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/e517a76c55704504aca04bbf57873a9f : FAILURE in 4m 35s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/09d7381870c74082b7675082c0e90bee : FAILURE in 5m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a81e97f236f14ee687562600a176b3dd : FAILURE in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5148f5767c294030bd52e3ba35ca7881 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58c356cd83114b8595d95328ed813421 : SUCCESS in 4m 18s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/97034e124a0f4fd88301e1ea31de975a : SUCCESS in 14m 06s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/add4e819ff184ccf8d206e7dfd970781 : SUCCESS in 14m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a099a9aa63c34282a85130bbb5521fde : SUCCESS in 13m 32s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cc3e0d4325c47e2b5ac317b883f4f86 : SUCCESS in 15m 29s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/eed764d051db4d658b976a66ba7c38b4 : SUCCESS in 16m 21s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b99a577019fc46caa84b23adfdf4c617 : SUCCESS in 13m 35s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6f406d56a6545d9b1827450b218dd12 : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":2},{"id":"0eb1e99accf941f9821cfec5184f84598d55f9b7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-07 09:18:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"70d9e36dfac762a000daa9c0b05ed69958064d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 09:44:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e7004896d94d4cc0952826377e1dff7c : SUCCESS in 9m 17s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/7673dcd7adbf4799add84d8798609425 : FAILURE in 5m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c2a4ebcc48e64f3fbcb50f7d3107f4de : FAILURE in 5m 36s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e27638df1574471ab4cb5fe6470bf31 : FAILURE in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e0fc89ca8e543c4a9ea9b483c51cb78 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52319adb0995427e8277c339904f530c : SUCCESS in 3m 44s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/c178982e2d0c44dba560c4cdefe14bfc : SUCCESS in 17m 22s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3111e837a33c431ba917ca833fbd47a7 : SUCCESS in 15m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b7e75edc6254f42aa4119d2a8bc7b4c : SUCCESS in 16m 33s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7475adeb479498dbe581ec3f8847a8f : SUCCESS in 10m 40s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/db6b7a74a54646ceb1d0ecccffd801a8 : SUCCESS in 13m 21s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ae37dc169b246b0992e7cd1bccba580 : SUCCESS in 12m 40s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50fab9eacd944597b1cec170b6655688 : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e61f930a0b502811becbbe11babca83cc923d77e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 13:01:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1d088e9f141540467b55d6382e2df4878f08bb43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 13:26:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cc8789a0c320466eb9e7bcfb3ae74073 : SUCCESS in 5m 25s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/cf21a3d699014addbdc0e226524a5263 : FAILURE in 6m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/af7a7f8ebd6042ecb9514d171db6c588 : FAILURE in 7m 34s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b634fb3cdd745bc85ab7d823d11778e : FAILURE in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169938cbc0df4d69b0be081477b5bece : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4573bdd057034d26acde1575267edb17 : SUCCESS in 4m 06s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/1330babc1a614b4ca7ab189293c74c95 : SUCCESS in 16m 03s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eacf3c89fe4b4d3885fa96a0374db2b6 : SUCCESS in 12m 07s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42b53d9daa7b4eb4b1518d2756457040 : SUCCESS in 14m 38s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0751c56c6fe4fe283e7d00ff7184de3 : SUCCESS in 13m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/163cdabd773044c4984c0a4573586c07 : SUCCESS in 17m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcd5991ace714e3ea15f30d2d6144c5d : SUCCESS in 12m 58s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62c5ffa48f434574a6a6e2f7a84ee848 : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":2},{"id":"9d9b0db8779974620eda3f0f04f9967adddc8817","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 14:16:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a334d3e6a382507769815f8f5a3d741d3f4fadc6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 14:18:18.000000000","message":"Patch Set 2:\n\nok, seems to be related to py2","accounts_in_message":[],"_revision_number":2},{"id":"b1867f97f6e6b42f1b39af707cd77152cfe582a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 15:14:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7091d643451344649b403d5f35207044 : SUCCESS in 6m 12s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/bf580c3fb782415aa7b4aa721dac9a1e : FAILURE in 4m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1ebdb790e078417aa8e82ea23b8ca4c0 : FAILURE in 4m 58s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c5a7a206ef346ba9e0fdc9949b42f19 : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9781806d46a94579a1de516d019aec60 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98f3d6e66a6d49ddb3cf254da34b67df : SUCCESS in 3m 18s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/cdc66b4a89fd4364ac3e44775c21dba9 : SUCCESS in 16m 15s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55c1608f2d1e45c58113f6803f8b12e1 : SUCCESS in 12m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c69a288073aa4745b8abe18ebed63a19 : SUCCESS in 12m 08s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7f091e45ba04516bd9f00ce8b62257e : SUCCESS in 11m 53s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/fde18fac46f14be79c6209ee97def079 : SUCCESS in 15m 40s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a52506bff6a54975b340e3273deb74f9 : SUCCESS in 15m 14s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9fc42d1513a43d38a9dfe251b98337d : SUCCESS in 11m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ca95c991c6092abaef74c0d6fab30975cce01922","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 16:18:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1258b70a5b08512af03e70895e74ddfc59f57076","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 16:18:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cac7c14384f3c5698892a8c42ab60d456552df5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 16:59:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b8b38f04496c4fa3a844ace8c5dcb29f : SUCCESS in 7m 17s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/760424dc845c447e8a0514aebdaddc55 : SUCCESS in 9m 19s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/bb864804b1fd409bb6b7cea3767f7205 : SUCCESS in 14m 20s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8343bd62f1641c58edac46f1caaef5d : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd7a590f3fb40d1a0c3a3b1d7973562 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60c2ad278efe484ca35b7db89c77d29b : SUCCESS in 4m 02s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/298a997d6ee54b99ad043a4bc9b6f366 : SUCCESS in 15m 04s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40d759728a74453fa6c93efef2cd795b : SUCCESS in 22m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5d16a084c1a423498feb9e85a41537f : SUCCESS in 12m 01s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50e350af27a3447698266ecb43ab510b : SUCCESS in 12m 01s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/6561617e90a74b3ab1f480929d7aef21 : SUCCESS in 13m 32s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6d2549bcef048b8b8da4fbb837f7103 : SUCCESS in 12m 13s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/290f33ca74d54fe6b32856d4feb427ae : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":3},{"id":"d03bed6b9bc7c1b4d7429aba2e58511095dcd1fb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-09 13:28:47.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"fe9b349e46dea8c8ecabac7884a4cfa38490f45d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-09 13:28:49.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"53b91874407f9fe4aa2353f0602a3cb76df6f4c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:29:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"53c44bd58edd6e4be53ea391879f1799576a7016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:48:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f419027e47ad86fe25f8b9060f1af59193aa4b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:48:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1037b01dc4624123811d5c324014af71 : SUCCESS in 6m 05s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/6ad1933f5d524dd0a8a7d3683bd16dfc : SUCCESS in 14m 58s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e7827753d0ac418a96d8f1118dabbc09 : SUCCESS in 12m 47s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76dd7c57ae8b446dab076ca641d8fe96 : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb20cf41eaf465089823b855370b1bf : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cd1491fb23440118f4247f740492c20 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"57c079459d7125ff4c53e46b0f0ca49193fa059b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:49:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ac65c82009466a88b1f57d68edfd1c : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d3d194d94e04d849e8b92aec203aa73 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1f1be2ee5a33ead547223c4770f70404397fde17","revisions":{"83b08f07da91d3e637823ef3fddb0004947b3eb0":{"kind":"REWORK","_number":1,"created":"2019-10-22 19:14:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/690342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/42/690342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/1"}}},"commit":{"parents":[{"commit":"23c6e8525986b0635f0d8c55a3757ec57fb1b29e","subject":"Merge \"PDF Documentation Build tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/23c6e8525986b0635f0d8c55a3757ec57fb1b29e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:14:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:14:31.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id051ad4936b4ed8f6a5c7eb3860ba793dc7fbbcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/83b08f07da91d3e637823ef3fddb0004947b3eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/83b08f07da91d3e637823ef3fddb0004947b3eb0"}]},"branch":"refs/heads/stable/train"},"701890a8fe0f981155247244d7213df632fca2c3":{"kind":"REWORK","_number":2,"created":"2020-03-24 18:32:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/690342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/42/690342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/2"}}},"commit":{"parents":[{"commit":"69103ff82ba7acaa6b51ffce96d1f67e2b398664","subject":"Merge \"Update centos7 lxc image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/69103ff82ba7acaa6b51ffce96d1f67e2b398664"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:32:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:32:43.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id051ad4936b4ed8f6a5c7eb3860ba793dc7fbbcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/701890a8fe0f981155247244d7213df632fca2c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/701890a8fe0f981155247244d7213df632fca2c3"}]},"branch":"refs/heads/stable/train"},"1f1be2ee5a33ead547223c4770f70404397fde17":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-08 16:18:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/690342/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/42/690342/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/42/690342/3"}}},"commit":{"parents":[{"commit":"69103ff82ba7acaa6b51ffce96d1f67e2b398664","subject":"Merge \"Update centos7 lxc image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/69103ff82ba7acaa6b51ffce96d1f67e2b398664"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:32:43.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-08 16:17:59.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nDepends-On: https://review.opendev.org/756883\nChange-Id: Id051ad4936b4ed8f6a5c7eb3860ba793dc7fbbcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1f1be2ee5a33ead547223c4770f70404397fde17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1f1be2ee5a33ead547223c4770f70404397fde17"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
