)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~690434","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ia6a4d27702ccfb80b0814e88e43c586ff9b0ec23","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"use-opensuse-15","hashtags":[],"change_id":"Ia6a4d27702ccfb80b0814e88e43c586ff9b0ec23","subject":"Switch to the opensuse-15 nodeset","status":"ABANDONED","created":"2019-10-22 22:47:20.000000000","updated":"2019-10-29 17:34:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2d7f49796f8d99394c201145125232363091f7e","_number":690434,"virtual_id_number":690434,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-22 23:07:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 23:07:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"159806144b3a97b1598cd3f6bb42d18313e1e445","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-22 22:47:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a08eca187bde6ca7c1ff2a7bccd63c2952c7f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 23:07:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/470cd5f9af964d93ae16e1c1d6fc5686 : SUCCESS in 5m 35s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/6c90eb1e28b24f5a992aeb3c87ec3ebe : SUCCESS in 7m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/558ca8b762ea46819cd228e275e0229c : SUCCESS in 8m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5274992a33154cc4a3da933699acf9a5 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98115b0cf20f4d259e0dc17aec571d19 : SUCCESS in 4m 21s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/109eef4647744d9ead78171881bcced8 : SUCCESS in 13m 55s\n- openstack-ansible-lxc-dir-opensuse-15 https://zuul.opendev.org/t/openstack/build/5acddcdc6fd548bd949451261046404d : FAILURE in 12m 25s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f05254c948b4424abaf72a7fc3b74114 : SUCCESS in 11m 59s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0d73655f1ba43f3983f96a91dfeeb2d : SUCCESS in 11m 16s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-15 https://zuul.opendev.org/t/openstack/build/7e5bf19af2084c918d7b0d5baa62f05d : FAILURE in 9m 16s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85aa7d963b4942f0830665e3d4e8c3fb : SUCCESS in 10m 17s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/bb76e8bb66314b219f1bfbafd70b0bd9 : SUCCESS in 15m 39s\n- openstack-ansible-lxc-machinectl-opensuse-15 https://zuul.opendev.org/t/openstack/build/a682b11fa70f4fbfaa53b8bd781f91ad : FAILURE in 11m 45s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/602f100bf5684965948e8d5a315f9465 : SUCCESS in 11m 40s\n- openstack-ansible-lxc-overlayfs-opensuse-15 https://zuul.opendev.org/t/openstack/build/6431906b7da14c6fb166f14059637dac : FAILURE in 10m 51s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ca7fee96c8847b3ae0db76d0d93d6d9 : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fdab25660c15e8d2a6ff0717864812ba8b2f4518","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-29 17:34:19.000000000","message":"Abandoned\n\nLxc doesn\u0027t work on opensuse so these jobs should be removed instead. There are different changes to do this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0b312cc6be606102cc092b1be5347463f6ab4f5","revisions":{"f0b312cc6be606102cc092b1be5347463f6ab4f5":{"kind":"REWORK","_number":1,"created":"2019-10-22 22:47:20.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/34/690434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/34/690434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/34/690434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/34/690434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/34/690434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/34/690434/1"}}},"commit":{"parents":[{"commit":"23c6e8525986b0635f0d8c55a3757ec57fb1b29e","subject":"Merge \"PDF Documentation Build tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/23c6e8525986b0635f0d8c55a3757ec57fb1b29e"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-10-22 22:46:47.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-10-22 22:47:19.000000000","tz":-420},"subject":"Switch to the opensuse-15 nodeset","message":"Switch to the opensuse-15 nodeset\n\nWe are trying to deprecate the opensuse-150 label type in favor of the\nopensuse-15 label. The reason for this is opensuse 15, like centos,\ndoesn\u0027t support its older point releases and instead expects users to\nkeep up to date on its minor updates. This allows us to carry one fewer\nimage in opendev and ensures we are testing supported releases.\n\nChange-Id: Ia6a4d27702ccfb80b0814e88e43c586ff9b0ec23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f0b312cc6be606102cc092b1be5347463f6ab4f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f0b312cc6be606102cc092b1be5347463f6ab4f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
