)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~690441","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I70789d5cb801d9b7658d515482688be97b948cc5","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"use-opensuse-15","hashtags":[],"change_id":"I70789d5cb801d9b7658d515482688be97b948cc5","subject":"Switch to the opensuse-15 nodeset","status":"ABANDONED","created":"2019-10-22 23:18:30.000000000","updated":"2019-10-29 17:25:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3869df2775835accbdb09d1be304fc9cf9a81d14","_number":690441,"virtual_id_number":690441,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-22 23:38:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 23:38:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"483d12bc85e00526ec692050675bffeca0bf8207","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-22 23:18:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1945fa491e92eea8286a9e3e0859f172f1d108a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 23:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/344fc1697c614f4fa9fc9695ae8388d8 : SUCCESS in 4m 12s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/c5b6aebbba714c2a80b032f02299f6fd : SUCCESS in 6m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/830293f5f9f94baab2ec9f522d858754 : SUCCESS in 13m 19s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5383dd3e19c3431aa027453db74be4a8 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3eb30673b1435b907266d4303aa46c : SUCCESS in 4m 42s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/55d2e5c36e134d0688648eead128434d : SUCCESS in 14m 35s\n- openstack-ansible-lxc-dir-opensuse-15 https://zuul.opendev.org/t/openstack/build/677f11db5d384e87a1ca63c8403cbabe : FAILURE in 8m 59s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f97e87a359c4e6bb12a7f5e4e8ac83b : SUCCESS in 10m 31s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d1ff65aed7546f987667a3bc0636e6d : SUCCESS in 11m 33s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-15 https://zuul.opendev.org/t/openstack/build/ffe7ce7cc82b4ccbb4f403d0af16ce59 : FAILURE in 11m 21s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b14badc001cc4fe0a9a6f1ab6ed983f4 : SUCCESS in 10m 33s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/26a1522768d44009838d0eae691d5e99 : SUCCESS in 14m 08s\n- openstack-ansible-lxc-machinectl-opensuse-15 https://zuul.opendev.org/t/openstack/build/6fab307823814d7aa6d83554fae1510c : FAILURE in 14m 03s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19dfae2b3f0e413886ab437268f49137 : SUCCESS in 11m 37s\n- openstack-ansible-lxc-overlayfs-opensuse-15 https://zuul.opendev.org/t/openstack/build/88402b3826f44903bb7e6e2eab33c8bc : FAILURE in 9m 48s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3e7572649e5435881f0c22808750d70 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f60e69ee2275640b579bd851ff9c7aaa4d299417","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-29 17:25:01.000000000","message":"Abandoned\n\nLxc doesn\u0027t work on opensuse so these jobs should be removed instead. There are different changes to do this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e99b791710a5e21e74614daf5b8b51973a8bc62","revisions":{"5e99b791710a5e21e74614daf5b8b51973a8bc62":{"kind":"REWORK","_number":1,"created":"2019-10-22 23:18:30.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/41/690441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/690441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/690441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/690441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/690441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/690441/1"}}},"commit":{"parents":[{"commit":"23c6e8525986b0635f0d8c55a3757ec57fb1b29e","subject":"Merge \"PDF Documentation Build tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/23c6e8525986b0635f0d8c55a3757ec57fb1b29e"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-10-22 23:18:02.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-10-22 23:18:02.000000000","tz":-420},"subject":"Switch to the opensuse-15 nodeset","message":"Switch to the opensuse-15 nodeset\n\nWe are trying to deprecate the opensuse-150 label type in favor of the\nopensuse-15 label. The reason for this is opensuse 15, like centos,\ndoesn\u0027t support its older point releases and instead expects users to\nkeep up to date on its minor updates. This allows us to carry one fewer\nimage in opendev and ensures we are testing supported releases.\n\nChange-Id: I70789d5cb801d9b7658d515482688be97b948cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5e99b791710a5e21e74614daf5b8b51973a8bc62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5e99b791710a5e21e74614daf5b8b51973a8bc62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
