)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~690443","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Frocky~If8b3b1359a64a3983496df6210aa2983f22b2625","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/rocky","topic":"use-opensuse-15","hashtags":[],"change_id":"If8b3b1359a64a3983496df6210aa2983f22b2625","subject":"Switch to the opensuse-15 nodeset","status":"ABANDONED","created":"2019-10-22 23:21:31.000000000","updated":"2019-10-29 17:24:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88863a4143e2e16f147b883592356f563945cf93","_number":690443,"virtual_id_number":690443,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-23 00:23:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 00:23:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23f5c3ddfe6d9b437bcf61f7388d96b742c36cac","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-22 23:21:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1abd88e4bb506d844e309cc483ece3f39fe13654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 00:23:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4893f40ed41e40d6a220e34b92c09d41 : SUCCESS in 7m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ae8cb477835b40d1bb838c6aaff14df7 : SUCCESS in 10m 12s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/4795baca24e14386b6221634f7376d91 : FAILURE in 7m 37s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89ecd618cb064889b515d0ca819d1bd3 : SUCCESS in 8m 30s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/cdb14a508c5446a494e57dbb2bdb58f8 : SUCCESS in 10m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6b4664cb0cfb406ca56d96815e508458 : SUCCESS in 4m 54s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/35dc5ab2b50848208aa5f7dc84b47e2f : SUCCESS in 16m 00s\n- openstack-ansible-lxc-dir-opensuse-15 https://zuul.opendev.org/t/openstack/build/a4c83841c139458bb3c3b0422f505767 : FAILURE in 8m 34s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f168f77adb7847ffa52e77a5d10b24ea : SUCCESS in 12m 37s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afa5a53253f142bf9572d9c3587df046 : SUCCESS in 12m 36s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9f219c233f0f452c8626c8356b281d00 : SUCCESS in 14m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85d03c6efa7642c397ebc2e6590e9b40 : SUCCESS in 14m 06s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 https://zuul.opendev.org/t/openstack/build/e9782234d5ae46ce9ff2752a55d1b596 : SUCCESS in 15m 51s\n- openstack-ansible-lxc-btrfs-opensuse-15 https://zuul.opendev.org/t/openstack/build/cc553f5e06c24b7cb3dfb0238c73ba46 : FAILURE in 10m 21s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/901609a4d7e144d89447af4c1c495d81 : SUCCESS in 13m 16s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fa4b976bfc5412b90ecc5a96ea5920b : FAILURE in 13m 15s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/e11e1697a421459bb8c058a19498101d : SUCCESS in 13m 40s\n- openstack-ansible-lxc-machinectl-opensuse-15 https://zuul.opendev.org/t/openstack/build/8af9894c24974bb88162d531bc4d768a : FAILURE in 8m 47s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/19451f7020e04186939492a6ab0856b0 : SUCCESS in 12m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13375185238f4c78a9418c116fbbf134 : SUCCESS in 10m 52s\n- openstack-ansible-lxc-overlayfs-centos-7 https://zuul.opendev.org/t/openstack/build/2578717c2ce844d5b7a88b9f5e309841 : SUCCESS in 14m 16s\n- openstack-ansible-lxc-overlayfs-opensuse-15 https://zuul.opendev.org/t/openstack/build/c0d535ed0002491ca1a72b0ced83ec81 : FAILURE in 7m 00s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/cba0bfd37a46438c918b308b72c8dd0d : SUCCESS in 12m 35s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3527530f999d46aeb7e760fb8d80093d : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"97ac440c3ee2ab0013db1a441cc9ba854d90b8de","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-29 17:24:45.000000000","message":"Abandoned\n\nLxc doesn\u0027t work on opensuse so these jobs should be removed instead. There are different changes to do this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e0d19d078fcd9b90412fb6d8dbe5829ea22b9f1","revisions":{"7e0d19d078fcd9b90412fb6d8dbe5829ea22b9f1":{"kind":"REWORK","_number":1,"created":"2019-10-22 23:21:31.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/43/690443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/43/690443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/43/690443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/43/690443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/43/690443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/43/690443/1"}}},"commit":{"parents":[{"commit":"4e4edc2b9f2c5f38066381a1f486100795b5a1b8","subject":"Remove openSUSE 42.3 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4e4edc2b9f2c5f38066381a1f486100795b5a1b8"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-10-22 23:21:09.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-10-22 23:21:09.000000000","tz":-420},"subject":"Switch to the opensuse-15 nodeset","message":"Switch to the opensuse-15 nodeset\n\nWe are trying to deprecate the opensuse-150 label type in favor of the\nopensuse-15 label. The reason for this is opensuse 15, like centos,\ndoesn\u0027t support its older point releases and instead expects users to\nkeep up to date on its minor updates. This allows us to carry one fewer\nimage in opendev and ensures we are testing supported releases.\n\nChange-Id: If8b3b1359a64a3983496df6210aa2983f22b2625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e0d19d078fcd9b90412fb6d8dbe5829ea22b9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e0d19d078fcd9b90412fb6d8dbe5829ea22b9f1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
