)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~690680","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I372d72922ad58abc46b816f953e68972770d808a","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I372d72922ad58abc46b816f953e68972770d808a","subject":"tox: Keeping going with docs","status":"ABANDONED","created":"2019-10-23 17:23:32.000000000","updated":"2019-12-20 16:25:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9a656a62a7cab043cd27e36692c2f933fafa78bf","_number":690680,"virtual_id_number":690680,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 18:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"890c474070b3e8b48ceb0f0f8ac681d60d2f8a8b","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-23 17:23:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f380ecea9a0f95d5a50cc3def306bdcab100f57e","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-23 17:39:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1554b5c52e2d61fe3a11c2050469bc95f288664e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 18:16:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/be94edc554dc4b2c85d7d091e9f74bb9 : SUCCESS in 6m 07s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/54df0b0f301840bd9894eb4b281cc14f : SUCCESS in 13m 40s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ef71815b909a42afb6d8a0446140d85f : SUCCESS in 9m 59s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2c4cfd703bc405a9353d6cffad42a13 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93105b1042a9481cb512e37fae66617f : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ab08270ab934eed810de7f873286a31 : SUCCESS in 3m 20s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/a9af2ff6a0f04ba0bb03ec2c7b1417a5 : SUCCESS in 21m 19s\n- openstack-ansible-lxc-dir-opensuse-150 https://zuul.opendev.org/t/openstack/build/b9360bc123d04a3ca2566af9fded1d46 : FAILURE in 27m 19s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d9d5662d4ef4f04914c18a4903c2f99 : SUCCESS in 11m 21s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c3ed0b77d034306b9d0f4fab0bdfebd : SUCCESS in 14m 20s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/efe6d1b0f34244ee8a18b596976855d3 : FAILURE in 19m 03s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a7102270c78426f9170a31ee643a74a : SUCCESS in 11m 56s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/8167c0fbdc0043db8710ca200dc6f8ec : SUCCESS in 19m 17s\n- openstack-ansible-lxc-machinectl-opensuse-150 https://zuul.opendev.org/t/openstack/build/e392f4144d564d8e9bf7ec9842e0b033 : FAILURE in 25m 46s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a32a47c0d5241ef8678f5db6f16fa88 : SUCCESS in 23m 40s\n- openstack-ansible-lxc-overlayfs-opensuse-150 https://zuul.opendev.org/t/openstack/build/c233345f89d44d4aa86236f18002c914 : FAILURE in 30m 20s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d031d6cd9244fa282e221fe4685bcd6 : FAILURE in 21m 55s","accounts_in_message":[],"_revision_number":2},{"id":"388151da3eb6e5903f32affb04219007829026f8","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-12-20 16:23:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9a656a62a7cab043cd27e36692c2f933fafa78bf","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-12-20 16:25:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf77ff09418843c1281dc61dbdedc3bad7e2f4a9","revisions":{"818d9795d27c4eff6b65d922b3dbc63bb50a2ff5":{"kind":"REWORK","_number":1,"created":"2019-10-23 17:23:32.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/80/690680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/80/690680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/1"}}},"commit":{"parents":[{"commit":"23c6e8525986b0635f0d8c55a3757ec57fb1b29e","subject":"Merge \"PDF Documentation Build tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/23c6e8525986b0635f0d8c55a3757ec57fb1b29e"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:22:59.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:23:18.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: I372d72922ad58abc46b816f953e68972770d808a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/818d9795d27c4eff6b65d922b3dbc63bb50a2ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/818d9795d27c4eff6b65d922b3dbc63bb50a2ff5"}]},"branch":"refs/heads/master"},"fb3721bc6e8b95710873b5b9ea809262c5089136":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-23 17:39:43.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/80/690680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/80/690680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/2"}}},"commit":{"parents":[{"commit":"23c6e8525986b0635f0d8c55a3757ec57fb1b29e","subject":"Merge \"PDF Documentation Build tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/23c6e8525986b0635f0d8c55a3757ec57fb1b29e"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:22:59.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:39:40.000000000","tz":0},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\n1. Sphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n2. Remove the unnecessary blank line for tox.ini\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nDepends-On: Ie837d376bb1a1d93367369d25140dd8a9d5ac14e\nChange-Id: I372d72922ad58abc46b816f953e68972770d808a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fb3721bc6e8b95710873b5b9ea809262c5089136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fb3721bc6e8b95710873b5b9ea809262c5089136"}]},"branch":"refs/heads/master"},"bf77ff09418843c1281dc61dbdedc3bad7e2f4a9":{"kind":"REWORK","_number":3,"created":"2019-12-20 16:23:59.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/80/690680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/80/690680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/690680/3"}}},"commit":{"parents":[{"commit":"964d418ba5ef7889af87179bb5af8a863144ceaa","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/964d418ba5ef7889af87179bb5af8a863144ceaa"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:22:59.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-12-20 16:23:50.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\n1. Sphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n2. Remove the unnecessary blank line for tox.ini\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nDepends-On: Ie837d376bb1a1d93367369d25140dd8a9d5ac14e\nChange-Id: I372d72922ad58abc46b816f953e68972770d808a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bf77ff09418843c1281dc61dbdedc3bad7e2f4a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bf77ff09418843c1281dc61dbdedc3bad7e2f4a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
