)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~691948","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fstein~I1a8b7ec942eb594c669c3feb0f2b6db159d8b909","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/stein","topic":"use-opensuse-15","hashtags":[],"change_id":"I1a8b7ec942eb594c669c3feb0f2b6db159d8b909","subject":"Remove openSUSE 15 jobs","status":"MERGED","created":"2019-10-29 19:11:27.000000000","updated":"2019-10-30 10:45:13.000000000","submitted":"2019-10-30 10:39:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691948-1572431955842-e7a1f455","meta_rev_id":"3e214b8c9fcadbcdbbbb79d9cc7137e8536bbb1a","_number":691948,"virtual_id_number":691948,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 10:39:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-10-30 10:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-29 20:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-10-30 10:02:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 20:37:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-30 10:02:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-30 10:39:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f451fb2b1bf975fa901135c06e2e24a1c59817b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-29 19:11:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b063048d3b763ba2a1a425c2d206e7b85e81c94a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-29 19:15:58.000000000","message":"Topic changed from stein/backport/689012 to use-opensuse-15","accounts_in_message":[],"_revision_number":1},{"id":"5c531cbde26aaa06a71f0e981935abc4a2d74038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:30:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b0c0beb9a0634cd09188b3a4a9284cd8 : SUCCESS in 4m 59s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/c6aa6e036acb4df39ad950519efdf612 : SUCCESS in 13m 15s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/18e86aef74d940daa669764fe342ebe7 : SUCCESS in 15m 03s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30ee3c26702149ada0701a8058a8e48e : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d07be56d7f8472ab5fddcf1c54b0ded : SUCCESS in 4m 20s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/feec533e608f45f4a44c020058f0cc8c : SUCCESS in 20m 50s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/324e6c6ef82a40a78288fe2774345648 : SUCCESS in 19m 25s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77bfaf24bde14af7a534e7947e0e2be5 : SUCCESS in 15m 47s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/124132e28fd84778baa438957b2a14b2 : SUCCESS in 11m 48s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/7bc10bcd30d54f559ccfeb628969d59e : SUCCESS in 16m 45s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73221f4ec97b4c548984d307c9376b9f : SUCCESS in 12m 57s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb7bdfe322ab48dc8a9adb95ca2bd718 : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ca29268053be4efa11c89bff7a0e59762954bc97","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-29 20:37:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f3ce98cec01e1de1967c917379de6146503288f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-30 10:02:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d45d38be4404be9754290f09b3ee28854e59ae35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:02:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e578cfc50edcaad3a2d3c8cdd400f577aacc816","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:39:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"80a923c475a44c6ecc5bf022a2f54b689939173a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:39:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2f5216b616d44e06965aebc74ea33acf : SUCCESS in 5m 37s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/abf3add48dd14f928d453d042162091d : SUCCESS in 8m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/66240c90206840fa8db22678d81875a6 : SUCCESS in 10m 19s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6af36bcd25e14fc39098725b784fe492 : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26fa6deb8e3477e87dcde2c646cc3d9 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3e214b8c9fcadbcdbbbb79d9cc7137e8536bbb1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:45:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70852a7bd6b439ca3b3dd40a57563fc : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fef5f363a026e8aec7ea36cc73dcf441e34a19de","revisions":{"fef5f363a026e8aec7ea36cc73dcf441e34a19de":{"kind":"REWORK","_number":1,"created":"2019-10-29 19:11:27.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/48/691948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/48/691948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/691948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/691948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/691948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/691948/1"}}},"commit":{"parents":[{"commit":"cce5c54363593615447c83ffbc90645e0aa48307","subject":"Add Centos GPG key custom url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cce5c54363593615447c83ffbc90645e0aa48307"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:21:17.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-29 19:09:24.000000000","tz":60},"subject":"Remove openSUSE 15 jobs","message":"Remove openSUSE 15 jobs\n\nopenSUSE does not support lxc, remove all jobs.\nThere\u0027s no value testing a use case that will not work.\n\nChange-Id: I1a8b7ec942eb594c669c3feb0f2b6db159d8b909\n(cherry picked from commit 7cdb2d1ac34f20b6369252c98c181f66b40d7667)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fef5f363a026e8aec7ea36cc73dcf441e34a19de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fef5f363a026e8aec7ea36cc73dcf441e34a19de"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
