)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~710251","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Iea8b0028c10526f6f5a1a934051181c963e743c2","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"Iea8b0028c10526f6f5a1a934051181c963e743c2","subject":"Ensure procps is present in debian container image","status":"MERGED","created":"2020-02-27 13:20:45.000000000","updated":"2020-04-27 09:37:14.000000000","submitted":"2020-04-27 09:35:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710251-1587980147675-593d93a0","meta_rev_id":"417b21713feed55d976275504fdcf5fbce0b51a4","_number":710251,"virtual_id_number":710251,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 09:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-04-27 09:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-04-26 21:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-04-27 09:15:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 06:35:34.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-04-05 17:51:11.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-04-26 21:45:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-27 09:15:22.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-04-27 09:35:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17ecffd83aef53b054d3a46f48e53af55f61b2d4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-27 13:20:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0a50060928bf791d86d0d399021e93372f5dcc1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-27 13:21:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06bfd93d500ef3f18879611a0a2f1d38a6898917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:39:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ffe0a14ae3df47739fa424d77ca89f60 : SUCCESS in 13m 33s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8629277aed7f413782db3ccf9eadc43e : SUCCESS in 9m 42s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3b0006330d774fef9b14141dc44274c1 : FAILURE in 6m 11s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac0cc55343c34c5f84adc3bb085adef6 : FAILURE in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6858e006a75e4769ab44478b9ef977ef : SUCCESS in 6m 39s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/810513f8e3774f0f95cbe671fc7b588d : FAILURE in 20m 25s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ca0609a0a1543328e1cc6caea4c2d84 : FAILURE in 17m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74f251be83df4deeb12d3e609abc2e7c : FAILURE in 16m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90fa64e2fb51430992e62bf76c3a9c61 : FAILURE in 15m 11s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/b8c84e7114f240e694e5f6577c3acf7f : FAILURE in 14m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f0b9c81858a475dbeeaf4cd3f1c7a13 : FAILURE in 14m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f3e503ec6384e30ae0752e309ed4de7 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":1},{"id":"41d96a20c4b84175b02514ce358c58b6cfc95b58","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-03 13:41:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f3a9d293f40ab44cea624c9b499f93f2e0a7208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 14:06:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/13e2e3103303411d935731562b72789e : SUCCESS in 11m 56s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/42b8de2af1e7433883c0068f65ba55fd : FAILURE in 8m 06s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/67deb1256a93415e8b7c4cec7d615998 : FAILURE in 9m 22s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82c94302b01148b2a419a4c329afd161 : FAILURE in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f491bc3f1a3b422fa6de8b536eaac81a : SUCCESS in 8m 04s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/76d090f1e01341548e2fe4e1bb9e2cc3 : FAILURE in 17m 39s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60ca1bf4a2194fbea1d6c7149ae0a05d : FAILURE in 14m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c07403d041cd413fbbadf507049fc4d0 : FAILURE in 14m 16s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd92db3d1705427eb336dc52283151af : FAILURE in 13m 15s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/f1c2266a07b8468085e454a2158b6fbf : FAILURE in 18m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db590f1ba319465082aabde8958646a1 : FAILURE in 15m 29s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78fd981479c945b1b588f9a8a3b9314e : FAILURE in 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ce718ccc9f9f7ef2a16b203ed46562cda2d4453c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-04 09:24:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d21426baffec814fc9d774a7ff93ebf77a40b829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:51:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0ae13c2235924af59c2d25520d012dad : SUCCESS in 10m 33s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/445207c4c72045abab385180c8bb39fd : FAILURE in 4m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4df02ea70289451cbcbe1718637610f1 : FAILURE in 5m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83338eade6da4eeb87d4dda836b94cd0 : FAILURE in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2024e5403c4723bb6a5a4a0244b361 : SUCCESS in 5m 12s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/2a61807a19b148c69c86ca9a74860756 : FAILURE in 15m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b89660890e3b4f3a822c5a40a43ee8c2 : FAILURE in 11m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/629fe0ca7c744210ba555e52003c442e : FAILURE in 13m 33s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f362e71f1db944ca9a5eacc7f6ce8bc0 : FAILURE in 13m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/bf0feaaccfab4b348beb096adcf5d8eb : FAILURE in 16m 34s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f096630eb2f495391505839b0537e6f : FAILURE in 13m 57s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a4d56fb04d0474f9c624599e56a2126 : FAILURE in 12m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1cc7c8aa3e5f1d3fe9e373d585962da389af60c7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-11 14:53:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f1a14a4a6abc005a4c25f021e32caf620cf9ce21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 15:23:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f53e62625735434c98fdfd8a7d12a9df : SUCCESS in 8m 43s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9e641f3cd4b64fd286a799b30bdcc287 : SUCCESS in 10m 20s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/620e3438139c4a5fa4d2f4baa46aad77 : SUCCESS in 13m 21s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acdbefc7d7fb4a078bc1f1f7ee05947e : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89dc7dbd26c49d8bf91c1f53fbbc1fc : SUCCESS in 6m 50s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/503c736f9a974cd5a0c2c726f4bd850c : FAILURE in 20m 12s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/476a3df816ea4c87b625c322ff94b48f : FAILURE in 17m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21a3c87726754539aece9c111c4e5490 : FAILURE in 15m 46s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b16a951690c40e487c96eb7ae880c57 : FAILURE in 12m 57s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/5847c31dabe54ef3a5a2a7e9f2c2594a : FAILURE in 16m 42s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5be5f13a53994ef7836a5b78de7b739c : FAILURE in 14m 10s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f39f094d14c84a3e8da32555f9a4c1de : FAILURE in 16m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d12a9970cd5eee84e19ddd66e3cf411c6adff00c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-27 21:32:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9c0f05e4b67dfa7f54656c950e919f040897a2e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 22:04:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/93a08d2ae68e4799b78439d712682e0a : SUCCESS in 6m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a6d1be8df7ab497c80aa74296c846c55 : SUCCESS in 9m 01s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8152ddecbda24d3dbeb3b4961f7f213a : SUCCESS in 15m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fb1ef2478c94a6a8fdd2a55a1f55218 : SUCCESS in 14m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8b7baadf6a4930a59d11d3a1841b17 : SUCCESS in 7m 08s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/edb64f159dec43ee8cf74b1228fcf2f5 : FAILURE in 17m 29s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfd774766fee4c14868fa6ab612fc6bf : FAILURE in 17m 49s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/599e9b52ebf340fa8a37a898be7f36ae : FAILURE in 15m 13s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a31f9547be094b929b3c4ad7816d24a0 : FAILURE in 12m 08s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/00a0e6ad86104ee7b83fcabee8568bf3 : FAILURE in 23m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a098dc60a8d4e889b5b63e646b66ea3 : FAILURE in 18m 05s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0818a9c62ec4fc8b0427d5f8c41b8da : FAILURE in 14m 50s","accounts_in_message":[],"_revision_number":2},{"id":"8cb0d6f9b4e58c05b885d900327afeb36a6353bc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-03-31 05:08:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ce258a67c05fee6128654917f9e0b3759abf6407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 05:38:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/65089574c8574ad5973b836387d5b05a : SUCCESS in 11m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/358a28fbaba946cd82cd9b383ff40abb : SUCCESS in 11m 57s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/66c6953628294456b37fe23c8c0a80c5 : SUCCESS in 14m 14s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/592012dccdef443b842769e31aa908f1 : SUCCESS in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05f7a2d1ea744df98a7ecd5e54dc09c : SUCCESS in 4m 06s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/7c87b7cbe50242b99148fd716fdec6b0 : FAILURE in 19m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3ddf2b3b44f4c84b2d6489673afaa41 : FAILURE in 17m 09s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f04d5a8b29cd4e8baa6fac96901abcd1 : FAILURE in 16m 39s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/751f64088f4b485aa5cf084522ef3bd2 : FAILURE in 15m 17s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/11cd3f3db30243ddb3f9edb4065692c3 : FAILURE in 15m 27s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86470f161693452780cd010a6bbf9ae6 : FAILURE in 11m 59s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5869c46a11a495ea7e638e9a95891c6 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":2},{"id":"3e643262274f7afca8adc5e9b9a82e9416884a97","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-04-02 06:35:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"725793a313a358b0eebd59a2f7bb63c57f9fa128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 07:01:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1db1d0f8b2b54263b611b0cdbec81ad8 : SUCCESS in 11m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ba3268ae48e1478bb444ffad482791ba : SUCCESS in 13m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4fd5dfe503e44cb485603b269b3cf09c : SUCCESS in 14m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93130e6777c3458b895fffa191195e39 : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2865e8097241ca8fe071edc2dc50a0 : SUCCESS in 6m 54s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/b46b3d716dbe4a3293322455c95b3421 : FAILURE in 17m 19s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f35dcf6d3934b778655a46c6483c051 : FAILURE in 16m 05s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/815ec471591a4fb3b8e1d1bde98134fa : FAILURE in 16m 33s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ce811689d2548be97cb75ba312fe40d : FAILURE in 15m 50s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/36be60cb442547d5bf9b8f6dafc14628 : FAILURE in 17m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30cbcfe721074691ba118557b282548a : FAILURE in 15m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b34fbd28382488a9b664cdef507fde4 : FAILURE in 16m 36s","accounts_in_message":[],"_revision_number":2},{"id":"fdca191ba557f7d9ed324a48de808b7abefc3535","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-05 17:51:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9dc5f75a5a37932e3fd2f48932d8f6fdbd520da1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 18:39:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1a3cebf3ecb146b08cfe66628d05feca : SUCCESS in 5m 42s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2f4fc3532bae4521bd592a74c342a208 : SUCCESS in 8m 22s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/93a7cd0196034d7a930eb18a5ff0fc96 : SUCCESS in 13m 43s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/733ce2aa5de6401d90b82b90e39b825e : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d43922ab38a44048c619ce63e65f4c8 : SUCCESS in 5m 13s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/da55b80634a047bbb68082c11a619a12 : FAILURE in 13m 49s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/175033d5ee3c43edb99e88d3abab476a : FAILURE in 12m 29s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe6a615604db450f8643c6a2b78fcd7d : FAILURE in 13m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae42c533801b402894cc82329b022911 : FAILURE in 13m 44s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/20c1e0a9da934fae8bda7d80d1070f1e : FAILURE in 17m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/827efe03498b46e09a9073eb9964dafb : FAILURE in 14m 09s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/641caec2a8a6434db5e4638a4803e266 : FAILURE in 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d96bcf90c87aa4db97f95a8da8377299108a3e06","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-23 20:09:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"adc43e6db2e167b305750fc41155670b3c1a497d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 20:32:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c1a4cbdd19e44fe49889ed5ec168d5f9 : SUCCESS in 10m 12s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6ed86a3946f84f1aaf474ef48d36c2a8 : SUCCESS in 12m 05s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/268db185dd694960b6732e08fcdabc64 : SUCCESS in 12m 01s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7eb5f28ddc5b41dbb32c50cecd0691c7 : FAILURE in 14m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed62329b19f4f978ef280540056be42 : SUCCESS in 6m 29s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/afc46e7fff26456ca4b99518ec270603 : SUCCESS in 19m 20s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6e4158380884f4fa31fb71ce937e350 : SUCCESS in 16m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d155aea9e2046349e65484858a77041 : SUCCESS in 17m 30s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96247c7f58934ddaaee60881bac2efa5 : SUCCESS in 16m 26s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/712aff3159914888ab63c067b0423e85 : SUCCESS in 17m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89d7d3fefa63434885a3fd5e3cf4b964 : SUCCESS in 14m 45s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0417b1204e5443d9dc2d2b77247ced8 : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ababa65b5695b6bd44d9c955f6adb1aca5d6b066","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-25 09:17:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"84c244b5d0b305bf69f047cfc226395396a4e9da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 09:37:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b14b5e474b78453bbdc9d97240787ad9 : SUCCESS in 6m 03s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/02e133dae8254b8c91afa23ffb83e798 : SUCCESS in 9m 17s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/dd96d02d240e4633ad46604c678f88db : SUCCESS in 11m 20s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59edb81a10654869959e3694a6fd9da5 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78977b1c5aad4be4ac795148bb953f34 : SUCCESS in 5m 33s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/dc78a8c2b26d44f3b5cc3cdd1f07759a : SUCCESS in 16m 34s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ab583c004d142ed83edb43c4a4f72cd : SUCCESS in 13m 31s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbbe594573ab4014807fdf912f5cb017 : SUCCESS in 12m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1397eda9af8f41d69fd756af008bd6f1 : SUCCESS in 11m 47s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/4ea3502d0016477eb5f0eeccac87f6a6 : SUCCESS in 15m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dca977c312b44f8dabc6f6b943fb3e6c : SUCCESS in 14m 48s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa7ac2290c49402980f0cab4eeee7e4f : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":3},{"id":"4672f0fe4bbadce92beaab02c841367f556b5f88","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-26 21:45:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"82d1f759a92416da0e6d44c1fa6e1bf9464111a1","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-04-27 09:15:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cb2c36269443a941ce222457fbb834e35e9bb1fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:15:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9d629a861a3c1a247a16851b3d5c0b997fec1e09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:35:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9645e4a18578443d889520b2cdd7ca85 : SUCCESS in 7m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fdd3802b76b7406796a5651a1ccda17b : SUCCESS in 8m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8ff4d00584374be7bdc56c66463b7a57 : SUCCESS in 18m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c733acbd39d94689b345f55202bcf373 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e87851f1c8ac41f5bc34bd02d0d6ab30 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4f50336890a794b0066f2e1087146e47ac04b774","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:35:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"417b21713feed55d976275504fdcf5fbce0b51a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:37:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23e920cfc374e588f88e28311a9c05d : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8614820045727354705522eaedd645bd49009714","revisions":{"7c18eba23060793c5063b780cfc57a582c017c6b":{"kind":"REWORK","_number":1,"created":"2020-02-27 13:20:45.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/710251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/51/710251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/1"}}},"commit":{"parents":[{"commit":"d3c1910cdcb37b2930149860d2fff7e67901edd2","subject":"Add py3 installation for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d3c1910cdcb37b2930149860d2fff7e67901edd2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-27 13:20:11.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-27 13:20:14.000000000","tz":0},"subject":"Ensure procpc is present in debian container image","message":"Ensure procpc is present in debian container image\n\nThis appears to be missing from buster lxc images\n\nChange-Id: Iea8b0028c10526f6f5a1a934051181c963e743c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7c18eba23060793c5063b780cfc57a582c017c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7c18eba23060793c5063b780cfc57a582c017c6b"}]},"branch":"refs/heads/master"},"99a29fba7cfcfeb443991e8d1bcda9cc600b94a9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-04 09:24:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/710251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/51/710251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/2"}}},"commit":{"parents":[{"commit":"d3c1910cdcb37b2930149860d2fff7e67901edd2","subject":"Add py3 installation for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d3c1910cdcb37b2930149860d2fff7e67901edd2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-27 13:20:11.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-04 09:23:56.000000000","tz":0},"subject":"Ensure procps is present in debian container image","message":"Ensure procps is present in debian container image\n\nThis appears to be missing from buster lxc images\n\nChange-Id: Iea8b0028c10526f6f5a1a934051181c963e743c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/99a29fba7cfcfeb443991e8d1bcda9cc600b94a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/99a29fba7cfcfeb443991e8d1bcda9cc600b94a9"}]},"branch":"refs/heads/master"},"8614820045727354705522eaedd645bd49009714":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-25 09:17:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/710251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/51/710251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/51/710251/3"}}},"commit":{"parents":[{"commit":"72bd7fa4541ec352f3d9dfc5631d4152fa632b91","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/72bd7fa4541ec352f3d9dfc5631d4152fa632b91"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-27 13:20:11.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-25 09:17:16.000000000","tz":0},"subject":"Ensure procps is present in debian container image","message":"Ensure procps is present in debian container image\n\nThis appears to be missing from buster lxc images\n\nChange-Id: Iea8b0028c10526f6f5a1a934051181c963e743c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8614820045727354705522eaedd645bd49009714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8614820045727354705522eaedd645bd49009714"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
