)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~710253","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ifc091d499523b19c4d8821dab83e9e2a2ea12e0a","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"Ifc091d499523b19c4d8821dab83e9e2a2ea12e0a","subject":"Move all tox environments to python3","status":"MERGED","created":"2020-02-27 13:34:57.000000000","updated":"2020-04-27 09:38:34.000000000","submitted":"2020-04-27 09:35:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710253-1587980148617-bbffc0ce","meta_rev_id":"c9ae3cb75e77417fe4b7b5dd357985a151d4df70","_number":710253,"virtual_id_number":710253,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 09:35:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-04-27 08:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-04-26 21:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-04-27 08:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 21:46:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-27 08:59:21.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-04-27 09:35:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b355f8f8923cd1373305f504ce0a061b4ba6383a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-27 13:34:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15b4c9444d2d1b9ea29892e8e35b55b5b2208845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:05:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b548661285dd42ddb41117d36cf61414 : SUCCESS in 10m 19s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/64426ff28adc40c19b4d820791f17e7f : SUCCESS in 10m 24s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4a14ac3996d945b58d00c84ef60a288f : FAILURE in 7m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ccdd7c6d28744d7998f40b61879426c : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e061acec3a04c73b2e05d027c4272b6 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a7035621d4d4ef1b3069f70c540bf38 : SUCCESS in 4m 49s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/7c0e4a69e0db4a50b6a12b8427f0f180 : FAILURE in 16m 18s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/510859681def41ceac2e888405695c66 : FAILURE in 12m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d969a8abaee41deaf4b181c28397188 : FAILURE in 13m 54s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/681590e257404a9f870d49a4e12dc3af : POST_FAILURE in 20m 54s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/a9b15ef0d4904717b74a5b5053b67a26 : RETRY_LIMIT in 14m 52s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/762c90d2a2a84409bd6ab18d407df399 : RETRY_LIMIT in 15m 52s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49921b455a9e471d8c54f91c51826b1f : FAILURE in 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ac79f3daf7a904e82e4450ae34160adfe5cd0360","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-11 15:37:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a2495c52ecdab524f141b31cd3df8e05722a0228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 16:08:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1857b33ba6844e308752ab7d9749d22a : SUCCESS in 12m 01s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cf9ee2250fd645dab8df5b8c9c58579b : SUCCESS in 14m 31s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c6a57037d7144ce49634edbce698952c : SUCCESS in 11m 19s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09ca71fb3a8445109dc8ae1c8c3a0cc4 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a09276c05744d9280f80bf14159ed73 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3a7823086764a8cb78568bb19e28889 : SUCCESS in 4m 51s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/f2cdd0843cba4bae967b48b4a7e355bd : FAILURE in 16m 19s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2494098d2f9b4e9f8236c90808718a71 : FAILURE in 18m 30s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c3e31aad70b4ab4b1c72ff89c64b2ec : FAILURE in 15m 27s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87b0459d2af947d183d54395efa7b823 : FAILURE in 17m 26s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/c2fe0f5d732841628f256b00f856a57c : FAILURE in 15m 51s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72c7469fa03d411fb86811396f534b93 : FAILURE in 21m 46s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73aede99554545129bb28be17f60732e : FAILURE in 24m 08s","accounts_in_message":[],"_revision_number":2},{"id":"25cb60b799905ce2ed8b1b223ae500ecb2a5d476","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-23 20:09:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bb44227c4f335127a5dc9ef39a8a018cf6d41d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 20:33:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7db206e152594727948b2421265adf6d : SUCCESS in 8m 57s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/22beac4979714a928bd12fe43b31ba6c : SUCCESS in 9m 17s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d078055d1a2a4dd187807829cea04a34 : SUCCESS in 10m 40s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fb4e036719c4652b517a30167e4328f : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea575e19517143c6a3079ffb25f344c0 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b00c50431304afda49fa514bdc8245a : SUCCESS in 7m 11s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/624cb11bfec445408420435cb12687bd : SUCCESS in 19m 53s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/495bb132f9bb476e94ede8b944038b27 : SUCCESS in 17m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbdd591e457e4b56890e7ccc7c68fcb2 : FAILURE in 15m 54s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f3853c9c95c41ddadf37320508608d1 : SUCCESS in 14m 53s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/72182dece60f44148dc85c4e9693fbaa : SUCCESS in 16m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52c4af5a45604fe3ab17a94c53536e17 : SUCCESS in 13m 20s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c4c0f4514004a5dbc1de15c79dd51f0 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":2},{"id":"4f276b0e784737a65c667ef857a639cd005320d8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-25 09:17:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d1de5cbf9814427361e120dfe4c0f1e5298d8e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 09:40:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9df826282e7644adbbaf19231f7058a1 : SUCCESS in 7m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f54f27c9ec344a0e84a88c5da8d49418 : SUCCESS in 9m 07s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/686b0582674f42af97a8b353fb569260 : SUCCESS in 10m 54s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bd6ac63d56a4b319193593a8e03085a : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11adaedeccbe4fcd90326abaa7e6c17c : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/234e7a4dfe6f4459b36fd7e68ad6a122 : SUCCESS in 3m 39s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/4d832e0dbc6a477c9b6149691a50a7ce : SUCCESS in 16m 39s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17db140fd62449bea73c157a3d5fbbfa : SUCCESS in 14m 43s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d8da889537748c4a7381d8204843b00 : SUCCESS in 13m 36s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6db8c2a8bcd142f39de99554c1d70065 : SUCCESS in 13m 21s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/57173e61ce44432884ea636b492f7dc0 : SUCCESS in 15m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/001700e9500844f4956bc775de7d40b8 : SUCCESS in 13m 19s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a27a603baa1441c7ab1bec514d2a7d90 : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":3},{"id":"5c9dfd4af76f950e27676dbf15f487495301bdaf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-26 21:46:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fd303aebe33e97db02b66a89e3ab35828e367dd9","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-04-27 08:59:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8a4e84d24472917445b31230a5ab68585048b075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:15:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6eea248f5d1fe08afe104424eb4c97fb981cb802","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:35:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9f2488c86f21acdb9542b72cdc3bbad0b14265ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:35:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/82402a18f1844d309814bff04d95e591 : SUCCESS in 10m 44s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5050a5b8a98946f2800d78756ded282e : SUCCESS in 14m 29s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/184b56eff557481a850e4fbcae70ab86 : SUCCESS in 17m 40s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27e8f8a89367473e9f482995e8f629a0 : SUCCESS in 15m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bacce6fa00a4d51a26d100ac5069b4a : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c5834362a7d4a0090cba947cf913a7a : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"c9ae3cb75e77417fe4b7b5dd357985a151d4df70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:38:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69952658c0374ab0b473ffa52f86c836 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19763e1cae9545949d4b0d90e4af7027 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a216dd962c79bd89a8dcfcb3af6990bfa6316fb","revisions":{"8193e4e36b60f9a8dcf674e4644ef0a8c5f4eecb":{"kind":"REWORK","_number":1,"created":"2020-02-27 13:34:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/53/710253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/53/710253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/1"}}},"commit":{"parents":[{"commit":"7c18eba23060793c5063b780cfc57a582c017c6b","subject":"Ensure procpc is present in debian container image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7c18eba23060793c5063b780cfc57a582c017c6b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-27 13:34:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-27 13:34:25.000000000","tz":0},"subject":"Move all tox environments to python3","message":"Move all tox environments to python3\n\nChange-Id: Ifc091d499523b19c4d8821dab83e9e2a2ea12e0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8193e4e36b60f9a8dcf674e4644ef0a8c5f4eecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8193e4e36b60f9a8dcf674e4644ef0a8c5f4eecb"}]},"branch":"refs/heads/master"},"820f2d917c27de124bb9463de32149e46df2a5cd":{"kind":"NO_CHANGE","_number":2,"created":"2020-03-11 15:37:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/53/710253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/53/710253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/2"}}},"commit":{"parents":[{"commit":"99a29fba7cfcfeb443991e8d1bcda9cc600b94a9","subject":"Ensure procps is present in debian container image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/99a29fba7cfcfeb443991e8d1bcda9cc600b94a9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-27 13:34:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-11 15:37:10.000000000","tz":0},"subject":"Move all tox environments to python3","message":"Move all tox environments to python3\n\nChange-Id: Ifc091d499523b19c4d8821dab83e9e2a2ea12e0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/820f2d917c27de124bb9463de32149e46df2a5cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/820f2d917c27de124bb9463de32149e46df2a5cd"}]},"branch":"refs/heads/master"},"7a216dd962c79bd89a8dcfcb3af6990bfa6316fb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-25 09:17:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/53/710253/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/53/710253/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/53/710253/3"}}},"commit":{"parents":[{"commit":"8614820045727354705522eaedd645bd49009714","subject":"Ensure procps is present in debian container image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8614820045727354705522eaedd645bd49009714"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-27 13:34:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-25 09:17:30.000000000","tz":0},"subject":"Move all tox environments to python3","message":"Move all tox environments to python3\n\nChange-Id: Ifc091d499523b19c4d8821dab83e9e2a2ea12e0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7a216dd962c79bd89a8dcfcb3af6990bfa6316fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7a216dd962c79bd89a8dcfcb3af6990bfa6316fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
