)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~718882","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I4b74670edd7b4ae9710dbb39deb82c0775a1ff66","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"ensure_apt-listchanges_is_not_run_in_debian_containers","hashtags":[],"change_id":"I4b74670edd7b4ae9710dbb39deb82c0775a1ff66","subject":"ensure 20listchanges is not in debian container","status":"MERGED","created":"2020-04-10 04:51:37.000000000","updated":"2020-05-21 17:34:06.000000000","submitted":"2020-05-21 17:30:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718882-1590082234029-a134e064","meta_rev_id":"7b35b9c35016862aa476ea9b9eae82eb7251b1a5","_number":718882,"virtual_id_number":718882,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 17:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-21 13:18:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-21 17:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-21 17:13:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 13:18:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-21 17:13:38.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-21 17:30:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85491dba653f79e65d1c694fe9b2f012a0089a54","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 04:51:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bb8253a1b76fcf0c57c8e66dbd3205fc604548d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 04:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7808ac2d3b0435796324b00b984cd33 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1f49146ef0103a56db29ea334c69cd1a9cfc50b6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-10 06:39:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d29b106a79ff56af1cde43cb696265528b742d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 06:52:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b36d139970245d497996dec9b28eca3 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d511c570e2ffe695a43860d77ed722912e91e885","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-10 07:09:28.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM but absent tests concerns me","accounts_in_message":[],"_revision_number":1},{"id":"e48e8a9d060b36fcc646afb0a812bd78b8662cd8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-01 11:25:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0a205dbd8817ef72f72113e0cbdc4acd11d69dd5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-01 11:36:38.000000000","message":"Patch Set 1: Code-Review-1\n\nRather than add another task for all distros should this be added here https://github.com/openstack/openstack-ansible-lxc_hosts/blob/master/templates/prep-scripts/debian_prep.sh.j2#L22 ?","accounts_in_message":[],"_revision_number":1},{"id":"2856f23e9929d104ed4b7bb916757da9a61a97e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 11:44:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/38f53bdc69d848559272a5ead816cb51 : SUCCESS in 6m 33s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9b9fba704666428cb474bfdff86ed357 : SUCCESS in 9m 25s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d3a34eb29acd4bdf8abcf6d7fed0d09c : SUCCESS in 12m 51s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94581015496744b3a1fba34249d30ae1 : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c768c82a114dcb815d118a733a5da4 : SUCCESS in 4m 52s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/7656785829634e3aae79d5009f7d04c3 : SUCCESS in 17m 35s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c42ccd832d7a44ba91ae6b49c36781c0 : SUCCESS in 13m 05s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5f868ade0484750bc51b159ef9d089e : SUCCESS in 14m 45s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25aaa0326254479ea14c3d8bd9663cc6 : SUCCESS in 12m 53s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/4256c5ad6c804864bc4a266c195e3832 : SUCCESS in 16m 30s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97c6b422add64a6e89e5171e977ccd89 : SUCCESS in 14m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71643213994e4842b3c348ace9b57799 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d905fee8f5f5c1e74c29610438d6fa4bb06c8e78","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-01 15:25:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"75a20135407c45e5003a22c14e8a66e5c8e47d2c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-01 15:34:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d358574f03207a16fb662ed447fb135021f7a4a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-01 15:35:49.000000000","message":"Patch Set 2:\n\nwould have liked to fix this myself, but whatever gets this fixed","accounts_in_message":[],"_revision_number":2},{"id":"c3e7577d9296dc515ba9745ed1570eaf9bb6e28f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 15:50:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e6e92744195b473e83d2dcc430a9d44a : SUCCESS in 6m 29s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a208a7055e484ffaaf61cb4a843708de : SUCCESS in 9m 16s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/07e4faa49e324436ad4e7308e20053b8 : SUCCESS in 11m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/781718bbbabd4684a16214e5ad5c1885 : SUCCESS in 14m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026880dd2b7740afb6135c06c7596b20 : SUCCESS in 7m 07s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/0de65b27aab84b1d8f37fd900e9be5c8 : SUCCESS in 18m 57s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbaf0a05a0394e27af3fd34ed364d90e : SUCCESS in 15m 23s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88a856470896424ba36516cba41f80da : SUCCESS in 16m 13s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90acadbe5f7d4c5cb253ab2d92348013 : SUCCESS in 12m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/ec34998c766c486786bd34b14a42db17 : SUCCESS in 16m 45s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/409db519eb374000903ce4493e0af59a : SUCCESS in 14m 30s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba6d6613152a4c6199f0a32cef9a3518 : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"76bf7a232b84f81b6f6d047dbd26cb31709fff4d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-01 16:58:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"38c141da679ce9e06e802e3135135ebd32aec423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 17:21:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0ae6d86c1255446d998dff7a0c0bd239 : SUCCESS in 10m 51s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0c7d8b122e574932aebbd1e573bc1ea0 : SUCCESS in 9m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e9b297dcc7f447b48e1a7eeeb89253b7 : SUCCESS in 11m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9d32fb2abc34b63a6cc6246d4063eef : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b66e044575224a2f92ec2c85a9430a84 : POST_FAILURE in 10m 32s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/f0614fe9d93a4effba6ba3f1cdedcdb3 : SUCCESS in 15m 36s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fc1e44c8edf4a11976fbf2bf20e8be8 : SUCCESS in 15m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4193be169c7c4f30bb1070f7499212e3 : SUCCESS in 13m 34s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94e8d41c91254dc990289acfcd8d9bac : SUCCESS in 12m 24s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/550e41902b7b4554a1daaebd5f15164f : SUCCESS in 15m 49s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/410bf628895040b08f8fa2a3cac9427b : SUCCESS in 15m 13s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9201504e999d45708f9748fd125dbc7c : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":3},{"id":"bd6b13135d6f5a99ce3f10b0eac36068b027e095","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-01 17:29:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a732b813098b83abb4389db65c8fef1d23ab9f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 17:51:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ba9dd09aed4049a29ea63b4340a00f1c : SUCCESS in 7m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/67bd4ae6a4ba4fe287e4a3c713d204df : SUCCESS in 9m 56s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ce3d0ad236b6411294356fe8fd1eeacf : SUCCESS in 12m 16s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e3f1d3689c84e91be53d205f9347d47 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75ad499c854841e7a73736a3706eaa63 : POST_FAILURE in 6m 42s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/3dc6f57c1c5e4e45bb1120bec65dd9e5 : SUCCESS in 16m 31s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b97eb9e4c80f4a7c90b713d4a0d08ee6 : SUCCESS in 13m 11s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e1105a07a48434aa34cf399b3b013af : SUCCESS in 12m 37s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/862ac5baeed049cd9ceb90f097f8da60 : SUCCESS in 12m 12s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/742fd812191c4e34bd1acce2cea2ade4 : SUCCESS in 16m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d061ad9111b8426cbe5e20f7cab05abb : SUCCESS in 15m 31s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/407dc452369b49258776f2c53e03efb1 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":3},{"id":"2b66d00b47aac094a21c11e5acd8f0d16f393c92","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-04 21:05:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"44ce9f7db93acb6be97e7e06d29b3c3eb0557e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 21:31:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5159aef245f84be38a512ce609edb0a9 : SUCCESS in 8m 58s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ffaf055326cf41c68d3b937a5c8edf08 : FAILURE in 10m 14s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a353ed01653849d6b4c08eff314eaf18 : FAILURE in 18m 52s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e59649a571b4c84b6b1c2ee7292e148 : FAILURE in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3d9e6ee61e4ac1b2b0a61e9341a4a6 : SUCCESS in 5m 48s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/7207d962a337452b9095da491a762d65 : SUCCESS in 18m 16s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e7a53fa2d9440598ccb391b8f89dcac : SUCCESS in 17m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd76102e73754a1fac252a70d13f2755 : SUCCESS in 15m 38s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6de415b311445d7a6b10fe21bac66cb : SUCCESS in 20m 00s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/96b8c1b2f0d94fcf92c34e4d710fcd15 : SUCCESS in 20m 45s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f78b7bd58784f218260e036b391f385 : FAILURE in 13m 53s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a4f2e2cfd694584834034988985d403 : SUCCESS in 17m 49s","accounts_in_message":[],"_revision_number":3},{"id":"209d2d18a7a653305df4b312bd6ccdf7146eacbf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 20:51:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9f715128057b20c7bdca93b533cbd8a81a3aedce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 21:24:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aa84477ce3d749a891550533df38a992 : SUCCESS in 8m 00s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c7ce683f6e304183aa2e28937b29e1ea : SUCCESS in 11m 05s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9b6b237444a54818b651d2b0f8b5174b : SUCCESS in 13m 26s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/35e902b860804c949d0bf7f8c3abeda2 : FAILURE in 9m 35s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58be05d30ae84651afa363524948fbb1 : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da975ddb6082468b98eb5b1126627099 : SUCCESS in 6m 52s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/b88e7913e3e145448f13b9223bb8c04d : SUCCESS in 19m 15s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afd8b329b8b14bbaa12a8f58bd51cb4c : SUCCESS in 22m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a442a3fd149347ec9a52163e053b189a : SUCCESS in 16m 18s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3668a8dfbd444352bee9d5c7b459ef72 : SUCCESS in 17m 55s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/45848872169c4c4e8412f694ae4327e6 : SUCCESS in 27m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecc59d7c80c24cfaa283e33d7994a6f2 : SUCCESS in 14m 32s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66f8a049c9b24a8f91187ecd3564156a : SUCCESS in 17m 43s","accounts_in_message":[],"_revision_number":3},{"id":"0e7fe02b0bc49cde1654dd796f89a656f7527fa7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-21 13:18:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f9d8f34c541cdf5e0e6b364853f7467919430ca0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 17:13:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3dd757a76f929c0142aff72d6f725d77f388c2b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:13:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4c56e8e862a5aedeead471c5bfde28561bfa4ef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:30:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4363b72f696e4b42a17cee8d33976376 : SUCCESS in 10m 29s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/18aac57e816944e481a6483b175b99ca : SUCCESS in 13m 12s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2551cce64ec34788b64fbdc10b820020 : SUCCESS in 13m 39s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32034932e4fe4b87b875e2302afec31b : SUCCESS in 14m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570c30565c5741e9b25ab56d33b4245a : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":3},{"id":"afed96d1ca75458c44a315677b6a2312cd9ea286","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:30:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cc03671bab116aca3cf8b90ff1b780ce526019bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:34:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1612b7a193f48f9af1b37b2c4e38091 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"40535a00d389fd5b67822c63a98eedd95b6cab93","revisions":{"9264ff89914f9e839e6bc573ad8ee62f4c6d45ea":{"kind":"REWORK","_number":1,"created":"2020-04-10 04:51:37.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/82/718882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/82/718882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/1"}}},"commit":{"parents":[{"commit":"d3c1910cdcb37b2930149860d2fff7e67901edd2","subject":"Add py3 installation for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d3c1910cdcb37b2930149860d2fff7e67901edd2"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-04-10 04:47:58.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-04-10 04:47:58.000000000","tz":-300},"subject":"ensure 20listchanges is not in debian container","message":"ensure 20listchanges is not in debian container\n\n20listchanges has apt call apt-listchanges\napt-listchanges is not in the base container so we get a failure\n\nChange-Id: I4b74670edd7b4ae9710dbb39deb82c0775a1ff66\nCloses-Bug: #1839535\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9264ff89914f9e839e6bc573ad8ee62f4c6d45ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9264ff89914f9e839e6bc573ad8ee62f4c6d45ea"}]},"branch":"refs/heads/master"},"f3a1a81ffa113f85ca61a38535515cf3b49dc85b":{"kind":"REWORK","_number":2,"created":"2020-05-01 15:25:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/718882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/82/718882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/2"}}},"commit":{"parents":[{"commit":"d3c1910cdcb37b2930149860d2fff7e67901edd2","subject":"Add py3 installation for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d3c1910cdcb37b2930149860d2fff7e67901edd2"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-04-10 04:47:58.000000000","tz":-300},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-01 15:25:44.000000000","tz":180},"subject":"ensure 20listchanges is not in debian container","message":"ensure 20listchanges is not in debian container\n\n20listchanges has apt call apt-listchanges\napt-listchanges is not in the base container so we get a failure\n\nChange-Id: I4b74670edd7b4ae9710dbb39deb82c0775a1ff66\nCloses-Bug: #1839535\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f3a1a81ffa113f85ca61a38535515cf3b49dc85b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f3a1a81ffa113f85ca61a38535515cf3b49dc85b"}]},"branch":"refs/heads/master"},"40535a00d389fd5b67822c63a98eedd95b6cab93":{"kind":"REWORK","_number":3,"created":"2020-05-01 16:58:39.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/82/718882/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/82/718882/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/82/718882/3"}}},"commit":{"parents":[{"commit":"d3c1910cdcb37b2930149860d2fff7e67901edd2","subject":"Add py3 installation for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d3c1910cdcb37b2930149860d2fff7e67901edd2"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-04-10 04:47:58.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-01 16:58:36.000000000","tz":0},"subject":"ensure 20listchanges is not in debian container","message":"ensure 20listchanges is not in debian container\n\n20listchanges has apt call apt-listchanges\napt-listchanges is not in the base container so we get a failure\n\nChange-Id: I4b74670edd7b4ae9710dbb39deb82c0775a1ff66\nCloses-Bug: #1839535\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/40535a00d389fd5b67822c63a98eedd95b6cab93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/40535a00d389fd5b67822c63a98eedd95b6cab93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
