)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~723635","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ia5aaa8f41172fe200d1d7ce0c7b6f26f7834e38a","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"Ia5aaa8f41172fe200d1d7ce0c7b6f26f7834e38a","subject":"Cleanup py27 support","status":"MERGED","created":"2020-04-27 18:12:25.000000000","updated":"2020-05-11 11:09:21.000000000","submitted":"2020-05-11 11:06:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723635-1589195183237-b16e9fe7","meta_rev_id":"bbe21ac78ee0827e54730808d155a3b91d62e947","_number":723635,"virtual_id_number":723635,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 11:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-10 20:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2020-05-10 19:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-10 20:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2020-05-11 10:41:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 14:44:30.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-10 20:51:09.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-05-11 10:41:31.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-11 11:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10ef0b0c0917bfa5d83f6db758eebb1c5ef98ee9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 18:12:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12e30976e6103160fcf3d4fc04c042eecc6edb06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:35:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fdab3be06d6348978a16968f31ef391b : FAILURE in 5m 00s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/989e7827e330448b893f79b9dc65ee39 : FAILURE in 5m 22s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5d5bb7f7cfaa4d3aab3996663f40df8b : FAILURE in 5m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33b43bb58638406bba6b886fa187a7c7 : FAILURE in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9987e1e3f642f3bef610a20beefa66 : POST_FAILURE in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc6143a36fb44e90b54f8c29747a797a : SUCCESS in 4m 18s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/c63906e17e2b4dac9fb2eb48e4a5e17e : SUCCESS in 16m 51s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd791c1f856840c6b79a6e7da0938e17 : SUCCESS in 14m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94cf7933a2d54b6db1077a74e5b7dce9 : SUCCESS in 15m 12s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bca369a91a464cf78db0e1d472d89b71 : SUCCESS in 16m 00s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/d68ab228b03546b4b5461808115968f2 : SUCCESS in 18m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/487c7192ad67461ba0920f6dff13ff8d : SUCCESS in 14m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9d4a03c00494278887824778810d176 : SUCCESS in 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9c0e58a39916f2eb0823e0a5fc743de7d09ca789","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 18:38:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d686a01edc046b05cdbd8bbb4397387f3f941f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:04:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3877dfff263a4c2998cd40012dac95e8 : SUCCESS in 8m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7848089b67d54380b0dc062a5b886974 : SUCCESS in 9m 21s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4e32c762c14b4fb190774ea9156e3d52 : SUCCESS in 9m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e54bf457c3ff44f3a896b42e671b7da7 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b649bf7c42034f1f808b6d499d1939d0 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8779bcacc1f847d8b6c4fd1f81bf89db : SUCCESS in 4m 32s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/98c17091bddd4a8a97f9f194568fc4ab : SUCCESS in 16m 16s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c3dbb1a67e743b5bb79927114dc6c44 : SUCCESS in 17m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0dbb549e3fea4b2ea83221d3e57bd8df : SUCCESS in 14m 55s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fee2f6684d04a5d9e816de7f81764ea : SUCCESS in 13m 41s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/a3b7867c59ce4516a0fa2471c7056c8d : SUCCESS in 24m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4936a64f77434138a5a0cccbe0b334d3 : SUCCESS in 18m 05s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/146d656a7b554ca4a7d9914296cbc3d3 : FAILURE in 14m 43s","accounts_in_message":[],"_revision_number":2},{"id":"93bafde0ae10a0dceb872934c7568f42cd6b474d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-28 12:20:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"36c4d51508e4e61672247de9710811f5bdedd344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 12:39:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ceb4d233de7f42dda2da2e2373262cc1 : SUCCESS in 6m 45s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/77aa5b98106744e99ff2bd1e582fc5d0 : SUCCESS in 8m 47s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c9c1a16905ba4c4b8efcca5c431df1bf : SUCCESS in 11m 36s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64067ff4955247ad8e7d4775b04a17f9 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c588060ba2b4e5e9879b90148e17949 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acd9cccd938c4585bcf88cba74fd14aa : SUCCESS in 3m 52s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/fec4ceddb4b6476c98e3eb78284923f2 : SUCCESS in 17m 01s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eab70c2f898b453bb5655882629e3c38 : SUCCESS in 13m 21s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5883d1884204993bf93ec6cb3af2818 : SUCCESS in 14m 44s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45943416c92943b3a0848913738a9753 : SUCCESS in 13m 42s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/abc1bc237870480e9a1a7fe1591e4100 : SUCCESS in 17m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c5e3f7976514081894c3254b48cb19e : FAILURE in 12m 58s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c6a2e169d5547fc9af79307fd2ea48e : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":2},{"id":"91dfb65a213c390663c14dacdfbb83c55e86c8ad","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-04 21:10:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9d536acfa86e98847efb7a78e23cc8d62b1245e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 21:47:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d396b4c36a5640e7b9d95da85e604374 : SUCCESS in 14m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/323f3b1c76f3461e9ce7793a14db3212 : FAILURE in 13m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1b13c0769fae44d584744a008b27a2c1 : FAILURE in 13m 01s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a6689690b5646298a1f9392ca0b70fd : FAILURE in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd842386fde345138695f5c8bf789a54 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/632ad34a935f45658bdf09e922088182 : SUCCESS in 5m 33s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/01eeda1bd51846a389c36fce444ddd19 : SUCCESS in 18m 38s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dace62f7d539415198b463447ada54b9 : SUCCESS in 14m 20s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cea2f8031974fdcb460855ecac4ec05 : SUCCESS in 15m 32s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ee2d1cb84da4534a7f2b6a05192afd1 : SUCCESS in 16m 34s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/d253d0e5b18c4eb7b4f9e4b52bb7005e : SUCCESS in 24m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b2263b769324af3b11c0cc8f2694da4 : FAILURE in 16m 46s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fc5236d4dce452a9e7c9addbae60ad0 : SUCCESS in 19m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c619081e61c0b561e5489475e24d05962a4141d9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-06 14:44:30.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1e2ba3afaa809ff16c2462b259fa2408f64e170a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 15:13:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6b192bf64b134453a0e5a6512d34927e : SUCCESS in 7m 55s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/544305c9ecfa4a98b20a39d3563319bd : FAILURE in 9m 24s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a2fb2fe4a0d34a109b3fe9060435c80f : FAILURE in 13m 15s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c44fe1cabd443688f0287c4ee1be6b3 : FAILURE in 14m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbe090828af471d997ffdb529e89c37 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/830c87d7eab344c6b845815690f5fb4b : SUCCESS in 5m 49s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/28aef2b01de9461bb80846f657e8b402 : SUCCESS in 22m 39s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc8d47905804423482348120334ba18b : SUCCESS in 19m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/246f6a630179400c8da0a39c4f356e0d : SUCCESS in 22m 24s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f72a95782f4445758759820cceeed8ad : SUCCESS in 17m 06s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/ff884a92826042ab82a81522fe80b87b : SUCCESS in 19m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c636d1d313642a7a04c7c18b1073fac : FAILURE in 14m 05s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc7420829b964e81a1808221b999473c : SUCCESS in 21m 08s","accounts_in_message":[],"_revision_number":2},{"id":"23fff10092d23dc5d304a208212ec4953cf372f0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-09 10:19:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4a705e5ba205795968b6253f383bad904230fcce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 10:42:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/abea2e4ec7114192938b129e4a15ea5f : SUCCESS in 5m 10s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/776d94d536694e09bd647a7918787a86 : FAILURE in 8m 42s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/657fb275104c457cac2700316fa3d41c : FAILURE in 11m 49s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f42e3cfcc164d18bccf04f1194dcf59 : FAILURE in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaaca7a4953a46fbba20b37c4ca7177a : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c04c23f658e64ec8ba789a930b9c59bb : SUCCESS in 3m 14s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/38a66295f8d64e58ad6809669a9e7da4 : SUCCESS in 17m 51s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38629982e6ba4fd89b25ca42fbd50f62 : SUCCESS in 11m 57s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc578aa1e12e4278ba202b3a74433d36 : SUCCESS in 13m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22cfe312160648b699dd8d21859d3e9a : SUCCESS in 11m 41s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/52c842597f314556ae35acd5da7dc291 : SUCCESS in 20m 59s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aeaea3016d984227aa4032d6548b71ea : FAILURE in 10m 56s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4e0e9cf384f442baab00b199ab85a0c : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6e490964e79e226a6792d1f1e924703549d4710a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-10 08:26:23.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2f0c4c981e9bb6c510e23574f683511e09ee12e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 08:49:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7a002a32048347fab0d78cf22b92fc25 : SUCCESS in 11m 12s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ced7150d62d743d696ffb42ef28a3b4d : SUCCESS in 9m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ef7418916c02419da1cee6b3ec4c5513 : SUCCESS in 12m 50s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b99064d851841afaad5417f2b2e7b88 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f11f7a6de3240e0acc0257da693409e : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ae618e9f6914368b3b420e269ee50ae : SUCCESS in 3m 02s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/ddd96244f74a486bb36b649513166e80 : SUCCESS in 17m 38s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c26f8e27ad34fffae94530f0f2aff01 : SUCCESS in 14m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/315f8b5f3de144f9b8e0dac82dcb229f : SUCCESS in 14m 27s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ca5998fec864d60b1e3ca12c7eb49a2 : SUCCESS in 12m 35s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/08f4da35f636433d88df90cae916faaf : SUCCESS in 15m 34s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff72202afa1640c98e1e584ce7da331a : FAILURE in 12m 26s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19f760b65e4b41509375c9bfd643776f : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":3},{"id":"342c3f942abad3bc714b54372652224ecd4e55d7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-10 17:38:56.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c547d4427e7a0785c16beade25058bf0253f3cd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 17:57:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9a8e9ab2c6ea4b0589c97734df0690b1 : SUCCESS in 7m 26s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d7394d74a5b04c838de56412fdbabaa3 : SUCCESS in 7m 14s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8267a1be310f4bc39f4d772072740a58 : SUCCESS in 12m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53840782f5be48bbb564eed8934ced23 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/168e6e82acbc40a485a00c65ebd65d73 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/746e857cfd6d43a8906773eff51a484b : SUCCESS in 3m 52s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/4bc4c5c1691a4dd28e3c50e97112ca89 : SUCCESS in 17m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35a48aa59d9e4ec9a7201e8a27340abb : SUCCESS in 11m 58s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36941ea89e864f86893837cb0898818b : SUCCESS in 12m 01s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b693cffec8964ba0a24bca60b3f54298 : SUCCESS in 16m 56s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/07b1919b84d6482a9fb8329a717438f0 : SUCCESS in 16m 07s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0d2e13574c14d3583566277cf6a2bf0 : SUCCESS in 15m 27s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3fb8430c8ba4daeb720c46650bbb292 : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":4},{"id":"9a169422b12b3cac0a179ee3e9d1d4900fde8f43","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-10 19:05:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0409e5ddc17b3a9ae2f912e8b094a47708f79af9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-05-10 20:51:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8d17265b792f9d17297bc6223de3c5f4153f4781","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-11 10:41:31.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"447f3984222153eee4542d804f36606d56c199b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:41:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c2ee4732dd1f1b8f719a045a490aa03e3ec06f1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 11:06:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f363fdfbc16b4ac79de7feb81ddb2381 : SUCCESS in 7m 38s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f734c37444824fed918d1d16bd2d4ab5 : SUCCESS in 15m 42s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/23ff6fdbf4dd471ba52c3910a656491a : SUCCESS in 13m 08s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b64e026db494d76a93378d7cccb0974 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97aace45f83b4d3db425ca26ff24d367 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc1bdd1b1e24448a96b90457fbe6b70 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":4},{"id":"ad9959b241fe3c9087983700365b594c1c9fc413","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 11:06:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"158244f5c4a842b9e45b88617da0d71c1b9bab59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 11:09:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11acb027a28440b0aa9be9008c3cd9ae : SUCCESS in 2m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10654396639f4a97b596aaf73c385666 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"57672cbdf50c1cd960d8271cddedcf1f5a7f67d0","revisions":{"0a78e845b82dd17d05a0664f8478f3894d7b41ba":{"kind":"REWORK","_number":1,"created":"2020-04-27 18:12:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/35/723635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/35/723635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/1"}}},"commit":{"parents":[{"commit":"7a216dd962c79bd89a8dcfcb3af6990bfa6316fb","subject":"Move all tox environments to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7a216dd962c79bd89a8dcfcb3af6990bfa6316fb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 18:09:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 18:12:10.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove setup.* files, these are not needed for this repo\n- Cleanup */source/conf.py to remove now obsolete content.\n- Remove install_command from tox.ini, the default is fine,\n  move constraints into deps, cleanup a bit\n- Enable warnings for docs building\n\nChange-Id: Ia5aaa8f41172fe200d1d7ce0c7b6f26f7834e38a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0a78e845b82dd17d05a0664f8478f3894d7b41ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0a78e845b82dd17d05a0664f8478f3894d7b41ba"}]},"branch":"refs/heads/master"},"c2a1a79486a497f70947df323ea2a4bfd3a8a46a":{"kind":"REWORK","_number":2,"created":"2020-04-27 18:38:04.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/35/723635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/35/723635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/2"}}},"commit":{"parents":[{"commit":"7a216dd962c79bd89a8dcfcb3af6990bfa6316fb","subject":"Move all tox environments to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7a216dd962c79bd89a8dcfcb3af6990bfa6316fb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 18:09:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 18:37:59.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove setup.* files, these are not needed for this repo\n- Cleanup */source/conf.py to remove now obsolete content.\n- Remove install_command from tox.ini, the default is fine,\n  move constraints into deps, cleanup a bit\n- Enable warnings for docs building\n\nChange-Id: Ia5aaa8f41172fe200d1d7ce0c7b6f26f7834e38a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c2a1a79486a497f70947df323ea2a4bfd3a8a46a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c2a1a79486a497f70947df323ea2a4bfd3a8a46a"}]},"branch":"refs/heads/master"},"bc95ac13e006943bfbe175287c11a51ee30f5037":{"kind":"REWORK","_number":3,"created":"2020-05-10 08:26:23.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/35/723635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/35/723635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/3"}}},"commit":{"parents":[{"commit":"7a216dd962c79bd89a8dcfcb3af6990bfa6316fb","subject":"Move all tox environments to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7a216dd962c79bd89a8dcfcb3af6990bfa6316fb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 18:09:41.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-10 08:26:19.000000000","tz":0},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove setup.* files, these are not needed for this repo\n- Cleanup */source/conf.py to remove now obsolete content.\n- Remove install_command from tox.ini, the default is fine,\n  move constraints into deps, cleanup a bit\n- Enable warnings for docs building\n- Correct ansible search/match tests for ansible 2.9\n\nChange-Id: Ia5aaa8f41172fe200d1d7ce0c7b6f26f7834e38a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bc95ac13e006943bfbe175287c11a51ee30f5037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bc95ac13e006943bfbe175287c11a51ee30f5037"}]},"branch":"refs/heads/master"},"57672cbdf50c1cd960d8271cddedcf1f5a7f67d0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-10 17:38:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/35/723635/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/35/723635/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/723635/4"}}},"commit":{"parents":[{"commit":"7a216dd962c79bd89a8dcfcb3af6990bfa6316fb","subject":"Move all tox environments to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7a216dd962c79bd89a8dcfcb3af6990bfa6316fb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 18:09:41.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-10 17:38:52.000000000","tz":0},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove setup.* files, these are not needed for this repo\n- Cleanup */source/conf.py to remove now obsolete content.\n- Remove install_command from tox.ini, the default is fine,\n  move constraints into deps, cleanup a bit\n- Enable warnings for docs building\n- Correct ansible search/match tests for ansible 2.9\n\nDepends-On: https://review.opendev.org/726645\nChange-Id: Ia5aaa8f41172fe200d1d7ce0c7b6f26f7834e38a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/57672cbdf50c1cd960d8271cddedcf1f5a7f67d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/57672cbdf50c1cd960d8271cddedcf1f5a7f67d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
