)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~727295","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I3cb6675cae785dc6d47915da2d6f1ac34a893a1e","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I3cb6675cae785dc6d47915da2d6f1ac34a893a1e","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-05-12 17:49:18.000000000","updated":"2020-05-12 19:28:53.000000000","submitted":"2020-05-12 19:23:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727295-1589311396238-299f2a5a","meta_rev_id":"2d7f315130fd0ba3604373ee1e54a5dc750eaa2a","_number":727295,"virtual_id_number":727295,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 19:23:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-12 19:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-12 19:05:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 19:05:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-12 19:23:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4045afb0212d4c0ec12dcc696fa08b135a7ece0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-12 17:49:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dae3185aa408c638fdbe3f5862334dfefb41f441","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-12 18:03:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"27b2e615ed5fe2e7dd0fc4a87c0d82326b67551e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:32:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/18788d0dfe6a4c2fb275f2a86e1ff372 : SUCCESS in 6m 45s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1af4755d642349e798f6eabc1f83ca7c : SUCCESS in 8m 52s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/29eb7656dce04c6aa513b0279f7a6a4d : SUCCESS in 17m 11s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f9f014a33bd480b87f39ecdf28f594b : SUCCESS in 14m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e635ca676424664adf2bb9e4a65b29f : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a014c27315241c6b6f9b63b5b3063fb : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e1a2bd6408042b7955693072d4599c1 : SUCCESS in 6m 39s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/f4a68dfe80be4fa294e1432e61ed6d4a : SUCCESS in 21m 03s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d1d09963dab43d2a4ea21d026cadc9a : SUCCESS in 20m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/928c839b1e9d4d59abb34029d1cfa785 : SUCCESS in 17m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abee74dd07eb412b9e7c2395f3875a9b : SUCCESS in 15m 17s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/8392dc3f996a49ae8a5abb8194ba8c80 : SUCCESS in 20m 29s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e7c6b5365794fca84380a1fa30be170 : SUCCESS in 18m 18s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65b9e65ed77649ae8e88232c7648a9c0 : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"fa3be13d0d91ab198b2317b380306f2027e25dde","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-12 19:05:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fe54acd9fef7db44b523e03910475d7d6f848d17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:05:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2be4171903dcdb3798600b7fd866aa0f9925d07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:23:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0c33e2a3c900ddae57238486e031f979ccaf838f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:23:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/868f8f0078be48be889f38f920fc981d : SUCCESS in 6m 46s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c1b3aab2737f40f183e61c3fdcab0f3e : SUCCESS in 9m 13s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1e2c2e88e3e94e53acbd8e02ade3e252 : SUCCESS in 15m 16s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45c135f781cf49bc8682ed21dcf9f296 : SUCCESS in 13m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/af5081dcacc94cee8f585f759b56445a : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4694de4736947cf9e5843fe6e61ff7d : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98afda955e014be78a9eb4e12698de93 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"cb4d731669d41d9486cc2d084f32bd3dfbf95d06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:28:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3011a64b144a759df9e75784f7bb00 : SUCCESS in 4m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f44ea437aef94e60a8f32aa7fcd5487a : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","revisions":{"7ee19948834effca6e1b1ada0a01edf3b3f94906":{"kind":"REWORK","_number":1,"created":"2020-05-12 17:49:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/727295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/95/727295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/727295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/727295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/727295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/727295/1"}}},"commit":{"parents":[{"commit":"288526d60d14e6ccd4600b588b667f01c902a643","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/288526d60d14e6ccd4600b588b667f01c902a643"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 17:49:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 17:49:17.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I3cb6675cae785dc6d47915da2d6f1ac34a893a1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7ee19948834effca6e1b1ada0a01edf3b3f94906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7ee19948834effca6e1b1ada0a01edf3b3f94906"}]},"branch":"refs/heads/master"},"f19c97983348ff7e0ab1b6453ad321a4988ab20f":{"kind":"REWORK","_number":2,"created":"2020-05-12 18:03:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/727295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/95/727295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/727295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/727295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/727295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/727295/2"}}},"commit":{"parents":[{"commit":"288526d60d14e6ccd4600b588b667f01c902a643","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/288526d60d14e6ccd4600b588b667f01c902a643"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 18:03:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 18:03:02.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I3cb6675cae785dc6d47915da2d6f1ac34a893a1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
