)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~728108","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Idb16322ed6554656a54214bfd4a1d5d3b92503d2","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"Idb16322ed6554656a54214bfd4a1d5d3b92503d2","subject":"[WIP] centos-8 support","status":"ABANDONED","created":"2020-05-14 14:54:02.000000000","updated":"2020-06-16 17:42:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd3e57bfe68f5050932f48d1316997ce54f77128","_number":728108,"virtual_id_number":728108,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-15 16:55:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-06-15 19:12:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 16:32:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-15 16:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-15 19:12:03.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"58128e89b45faff17cd366d0efdce963623aced7","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-05-14 14:54:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dba48f6396dd9f6cf5084611d8ae6159619c1e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 15:19:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9fb7d64edbfc41ca9a5c88926a4e5f92 : SUCCESS in 11m 53s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8c23c18c6eb84403ae062b6724aaa8b2 : SUCCESS in 9m 25s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6c9a0310c0384257b7f1d351c3c645af : SUCCESS in 12m 48s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8b15aeceb09347148662aabdbeacb7ca : FAILURE in 8m 07s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53d0133db9f44a9a8ecdbadb29c4bd98 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7edf545c5a40faa64bd674bd028580 : SUCCESS in 6m 47s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/cd808006fd5940ada1a94f4ec2add201 : SUCCESS in 16m 39s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e6e6f670c6642d9a765b5300731f094 : SUCCESS in 14m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc21ac0fd2f8456b94a1a7486cb23820 : SUCCESS in 13m 39s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abeec7240d434390a9e5a06999ee0be8 : SUCCESS in 14m 50s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/92c3c451ccb64e339a2d84b5207e100c : SUCCESS in 16m 15s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3adbf7ed45a849dcb3ca0264585ee973 : SUCCESS in 19m 04s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/083c5879d8304af088db019dfc0c2395 : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e2121165569fcf5282f6ad908c6a8f0150614e71","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-05-14 19:13:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4da832595eabb72f2466f846c93e31a0621eda1d","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-05-14 19:14:57.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9dd9172ba948cfe2bf7017b594eaa84c6485646f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:30:50.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"3384155dd48a61364b6fdda885253d01bd5c1f16","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 12:03:35.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"7661d7d7e7a90923356829d7f4d0cabdaa8db031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 12:29:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5b5ef9daa647497080aeb187407b4eba : SUCCESS in 6m 44s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/68718b2e5cad492b81cd4a08f4c3d526 : SUCCESS in 15m 20s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1ee1de985d88493fac34c0076a41b069 : SUCCESS in 22m 23s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/57d69e13945e4fd6831b3c2c583c3338 : FAILURE in 13m 59s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5bf108b181e42b48143acb2cc0343ac : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f62014bdb384efbb059e40c960255b5 : SUCCESS in 7m 01s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/8b85d6bb88c84d379f12bba4f649f3bc : SUCCESS in 25m 11s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab2fd04506404e1e8b3b22bf93c433cb : SUCCESS in 16m 03s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e04b8b6b8b5486a919a007ed27f0a9c : SUCCESS in 17m 06s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b372fb3c4a9406cb8a2eba48b602c84 : SUCCESS in 14m 47s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/e8d29be79d2c47a9ba5b7e63ff52e80d : SUCCESS in 20m 30s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c1a22fb5d034f73b626b19512d72d16 : SUCCESS in 16m 59s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5d09b19b3164d329f8d6cd90396a7a1 : SUCCESS in 16m 57s","accounts_in_message":[],"_revision_number":4},{"id":"b7ad090609d409e3b6319f3932321373ae8b611c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 16:32:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8d8e4de8d6405a64a0aaf607d9068e05d2499c95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:00:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5f1f3cf9326c41288d02134320af14b7 : SUCCESS in 8m 35s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7f35b99ffceb43c88736849c017db4b6 : SUCCESS in 12m 18s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0d74e71c1711459fb801666ed54dd497 : SUCCESS in 14m 36s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/945fe2d809854b55ad801a9730b1974d : FAILURE in 10m 53s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46861189b6544caaa746b63b7a14e1c2 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3638df440de4447997b3ae0b193151e7 : SUCCESS in 7m 01s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/ff7f01f85caf4718b190b929f0e9be87 : SUCCESS in 19m 19s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bea3ff035eb47eca4f4eadb042dc675 : SUCCESS in 15m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c986e00317a74eebbf26d3b661dfbc9c : SUCCESS in 19m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/580f3ff6628243329053160aa994f3aa : SUCCESS in 18m 35s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/3663ab2c28664832a76225d125cc7a19 : SUCCESS in 18m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4496e3303d4a40e686badfcdcda4e0f5 : SUCCESS in 17m 06s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7348bb59d06344c4bb0bd03a882f1079 : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":5},{"id":"202fef16e80bc0b41d5fd89ef7129e998c65b212","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 17:33:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"18844d136b6e94c4545b1d418046f460028f432a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:02:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f810395b185340f1ba81348d91e307e2 : SUCCESS in 9m 12s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7b36088b5ea74e6c900322baec629960 : SUCCESS in 12m 32s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6528100592a24d1dbcab21c4cec20eaf : SUCCESS in 15m 54s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f03b9efafa604ecd90e3043443fc14d2 : FAILURE in 11m 03s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd71827d5f1541f7bf6842f7ab950a31 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ae17e8f7a44f568196756f17ff3b7b : SUCCESS in 7m 40s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/cc2f442384364a96bb5c89318803f239 : SUCCESS in 21m 55s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da79dc19ea25460aa26da1f5754fba1b : SUCCESS in 16m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24dc85d5b76746cf9c3e819b20115298 : SUCCESS in 21m 19s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97d9ef5eeead4a94a0d2467f7f3d8a36 : SUCCESS in 23m 08s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/bce07bf86790407ca321f6af1b613aff : SUCCESS in 21m 21s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3a281be14fb412694186ef1086291e7 : SUCCESS in 15m 27s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53e6c0159f234cb498a88b7112250688 : SUCCESS in 15m 42s","accounts_in_message":[],"_revision_number":6},{"id":"6e33fd4eb96db904e6f9b59a81300e6a6be6b6e9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 18:33:28.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"94e6741f1acf757635df26d2ba2192f98a7c2eb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 19:35:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4206e6c9843c459cbbf760127f2ee8fe : SUCCESS in 11m 26s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4feff933404643acae69d01f98428a88 : SUCCESS in 12m 16s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f8410139e451409cab36ddaba64e319e : SUCCESS in 11m 47s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/be6bee973ded4efdba190e8645f012d2 : FAILURE in 8m 34s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf00349e295b4e38885ed34c69c48c6c : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91018c01f6164f9a9e8fa92054b00589 : SUCCESS in 9m 34s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/239a2e81b02648fa9dec5837fb728a4f : SUCCESS in 19m 55s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/643e932cd4cc4b00bf4c72881f6aa1a2 : SUCCESS in 13m 53s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cb9b4efc63141d3917b09d3a0d33bfb : SUCCESS in 16m 19s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/887e271a767c40aba890ce5871b9e102 : SUCCESS in 12m 29s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/5a2f4b30e80a4f13ae6d666339101af5 : SUCCESS in 23m 12s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/463ed72fe1304d36a5218af25a91db64 : SUCCESS in 15m 03s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bffb08d5a33c4613b00c845338642c4c : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":7},{"id":"d9a814c7df9fd72b94927e2c3d0cef2700b66c53","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-05-28 11:52:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2a224a8613b9b4921ccb33d0d056b7bd611453df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:17:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5b576e6dc26e4b95adf11efab6a0d7ef : SUCCESS in 6m 05s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7aef94fa56524eaf959a45fbd1ef7dea : SUCCESS in 8m 15s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/647e4ff2c9f64d2283aab4eb4decb796 : SUCCESS in 12m 48s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8c4a7d6efdee4de3a65cb864a5dc7ca5 : FAILURE in 9m 22s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a5c79abc75649a88384f4dac8b5f83d : SUCCESS in 9m 27s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f56f085c4e104ff28ca2c9b38a04a280 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533cb2f76a96413ab02b8592942de7fe : SUCCESS in 5m 59s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/e1c50780a9a24c2c952d243068e4d961 : SUCCESS in 14m 12s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09c7f638919442209d0e10f15db85124 : SUCCESS in 14m 44s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c77a75af7fe94077bacf941f558295a0 : SUCCESS in 14m 57s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7253618a41354992b8de5b19e4ae537e : SUCCESS in 13m 35s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/1e30a9034ff84bdab0e7795ca262300d : SUCCESS in 21m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4413eb46c0af480ea5a523827feaedde : SUCCESS in 13m 06s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f08ca4d0bc04386874cf148f332b1c6 : SUCCESS in 13m 44s","accounts_in_message":[],"_revision_number":8},{"id":"14e562461aec50058cdf3c66b8005e67ede87fe6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-28 18:24:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c0a44390cdaecff1bdb2687144015d120fd7037b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 18:49:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dab07975979f471388e9de30255f2a31 : SUCCESS in 7m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8e07ab2ec5094b269e0c2e5b01f88d7e : SUCCESS in 10m 03s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5e3b92e854dc40bbbd92d777a5d2f4b6 : SUCCESS in 14m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/91cc87d7cdfd4746826ae002b0bf2d4c : FAILURE in 10m 55s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55f3ae1a16bc43d5b4730d24ee422040 : SUCCESS in 11m 39s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c60693fbada1416489b8a54fbc55bcb2 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ca03ee7cd34e67acf8a67d82a6decb : SUCCESS in 8m 12s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/2f317e34afa24fa586812f159d476ee3 : SUCCESS in 17m 08s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ba8b440661e4a18bca55400cc26f524 : SUCCESS in 16m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e330698db1147cb842a2417604ef19a : SUCCESS in 17m 09s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/790074e7161043089e566d54d3e96963 : SUCCESS in 16m 24s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/b3cbc103430642abb1313209063de0ba : SUCCESS in 19m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccb20b3e3a4647f4b5fb6a0660f511ac : SUCCESS in 16m 57s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c0b659a13264b72b53cdcf10795ada2 : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":9},{"id":"68fe47a437922eea44e8a9c0dfda5907d13d3ee3","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-05-28 19:13:01.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2cbe117abc93db97bd5a8bceb61621fdd3a55a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:48:18.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eb1355d4970f4b0d878d53cab957b1de : SUCCESS in 12m 25s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/746efb9bc1854c129bf77004344ca5b8 : SUCCESS in 10m 03s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ec53f9ae8ba44cbfa4795be3eb91c46c : SUCCESS in 9m 18s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/abdcc7887491403d8a1809179c5588d9 : TIMED_OUT in 1h 32m 28s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/231d225df0d44e9dbdf449400ce068fa : FAILURE in 12m 23s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fa7b08d68eb417db55b2af8f41e99b5 : FAILURE in 17m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750de0f5c8bf4bf5825f9bb2fecb2ebe : SUCCESS in 10m 18s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/141b65ce89204232877f5f1b11497982 : SUCCESS in 24m 28s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ace77e07de34e54aa835bfa6833b18d : FAILURE in 9m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dcbfd6d075a4c3889748153345db31a : FAILURE in 12m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdd131ee77f44d8ba6280221fd76fc73 : FAILURE in 9m 55s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/4e8659be90a740dfbf2ccc0e6f46b9b2 : SUCCESS in 16m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6455e1a3f624d728617958874e58426 : FAILURE in 7m 54s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a4c3031554c47a6be86ee4aa3ae5bbf : FAILURE in 8m 55s","accounts_in_message":[],"_revision_number":10},{"id":"857847d661952f3f8c3ba55208bfe1d0a85fb0b5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-15 15:13:33.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"efab6812acee5a4db4fa20a756a5e6d4eaf3d637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 16:55:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e2aad7331c674a8dbfe0d4c5da652c19 : SUCCESS in 8m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b7007ee8bb9f47d49e23194a0b4ad930 : SUCCESS in 11m 16s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a6dc9afcdf514775a832636024d252c6 : SUCCESS in 18m 22s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/23ab2777b62047d8b4685e3139a1b110 : TIMED_OUT in 1h 32m 40s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39ba6113f8de4c56b5127ed19d88ab19 : SUCCESS in 14m 06s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a9edfdee0bb4e5a85cafe41346fe41e : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d55ee12a0364f5fa066ff7279301778 : SUCCESS in 7m 25s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/fd25c135b89e4a828d119f2f3448b7bb : SUCCESS in 23m 34s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc13714c7cd54d71b5df3092b5588d2d : SUCCESS in 18m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a2566a2a9504d29ad688df74d598d4c : SUCCESS in 19m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5385d78ef049445fb2ee61392f4d3193 : SUCCESS in 18m 43s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87c0cf7fda5741b3b40e24b84f0fc98c : SUCCESS in 26m 52s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4368c91c95b6487d92cba21b4f7375e7 : SUCCESS in 18m 13s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32c5f901563c456b801ebc05bc007cd1 : SUCCESS in 18m 09s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/9fe9b99b96654fe99b827f9b1e9d635c : SUCCESS in 22m 12s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61c6a689883e4ccb9629b2577c49a32f : SUCCESS in 22m 02s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ff7bcb73d0d43b39c227dbebd12e43f : SUCCESS in 21m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff75285756eb48a98a3479d33a321af5 : SUCCESS in 16m 34s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7a25227f5ed496a93b7f1b302ac05a2 : SUCCESS in 25m 12s","accounts_in_message":[],"_revision_number":11},{"id":"a7d9cd4f8a75ec7be62b939e22c42d350420c242","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-15 19:12:03.000000000","message":"Patch Set 11: Code-Review-1\n\nI took the liberty of merging the content of this into https://review.opendev.org/#/c/735656 because it didnt make sense to separate them.","accounts_in_message":[],"_revision_number":11},{"id":"25a58e86d88b46c1ce2c05406946a86439128e14","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-16 17:42:35.000000000","message":"Abandoned\n\nmerged in another patch","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"e484938285960d2e4b727f9b0561b18bf5c3b748","revisions":{"badc9dbcb2ad2c75739aa526e0f15df5d0080753":{"kind":"REWORK","_number":1,"created":"2020-05-14 14:54:02.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/08/728108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/1"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:54:02.000000000","tz":-240},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nDepends-On: https://review.opendev.org/#/c/709454/\nDepends-On: https://review.opendev.org/#/c/709455/\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/badc9dbcb2ad2c75739aa526e0f15df5d0080753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/badc9dbcb2ad2c75739aa526e0f15df5d0080753"}]},"branch":"refs/heads/master"},"7cfee65be9f0fc9e2c2cfb082f00dd09b3c265a5":{"kind":"REWORK","_number":2,"created":"2020-05-14 19:13:27.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/08/728108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/2"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 19:12:11.000000000","tz":-240},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nDepends-On: https://review.opendev.org/#/c/709454/\nDepends-On: https://review.opendev.org/#/c/709455/\nDepends-On: https://review.opendev.org/728259\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7cfee65be9f0fc9e2c2cfb082f00dd09b3c265a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7cfee65be9f0fc9e2c2cfb082f00dd09b3c265a5"}]},"branch":"refs/heads/master"},"47bdc4bed87b17190b66d6d5fd2f5c169fcff9dc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-14 19:14:57.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/08/728108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/3"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 19:14:25.000000000","tz":-240},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nDepends-On: https://review.opendev.org/#/c/709454/\nDepends-On: https://review.opendev.org/#/c/709455/\nDepends-On: https://review.opendev.org/#/c/728259/\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/47bdc4bed87b17190b66d6d5fd2f5c169fcff9dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/47bdc4bed87b17190b66d6d5fd2f5c169fcff9dc"}]},"branch":"refs/heads/master"},"8789765a214829607fd3727b101b38b180bafbd4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-26 12:03:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/728108/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/4"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-05-26 12:03:32.000000000","tz":0},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\n\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8789765a214829607fd3727b101b38b180bafbd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8789765a214829607fd3727b101b38b180bafbd4"}]},"branch":"refs/heads/master"},"d734f1ff008d708049c4757c690e2b16a41978a4":{"kind":"REWORK","_number":5,"created":"2020-05-26 16:32:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/728108/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/5"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-26 16:32:44.000000000","tz":180},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d734f1ff008d708049c4757c690e2b16a41978a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d734f1ff008d708049c4757c690e2b16a41978a4"}]},"branch":"refs/heads/master"},"45cca4cf2890585a238a101311c4fa3920dac35e":{"kind":"REWORK","_number":6,"created":"2020-05-26 17:33:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/728108/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/6"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-26 17:33:06.000000000","tz":180},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/45cca4cf2890585a238a101311c4fa3920dac35e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/45cca4cf2890585a238a101311c4fa3920dac35e"}]},"branch":"refs/heads/master"},"8b0178ec4c79d7527e1b6614a6ee4cb415d0b5af":{"kind":"REWORK","_number":7,"created":"2020-05-26 18:33:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/728108/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/7"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-05-26 18:33:23.000000000","tz":0},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8b0178ec4c79d7527e1b6614a6ee4cb415d0b5af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8b0178ec4c79d7527e1b6614a6ee4cb415d0b5af"}]},"branch":"refs/heads/master"},"5ea75c46a85f952daeb693bbaae485df0807624d":{"kind":"REWORK","_number":8,"created":"2020-05-28 11:52:01.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/08/728108/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/8"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-28 11:52:00.000000000","tz":-240},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5ea75c46a85f952daeb693bbaae485df0807624d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5ea75c46a85f952daeb693bbaae485df0807624d"}]},"branch":"refs/heads/master"},"654d5dfe1348218e2e4d774a8c2a5deab5cd5c90":{"kind":"REWORK","_number":9,"created":"2020-05-28 18:24:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/728108/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/9"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-28 18:24:05.000000000","tz":180},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/654d5dfe1348218e2e4d774a8c2a5deab5cd5c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/654d5dfe1348218e2e4d774a8c2a5deab5cd5c90"}]},"branch":"refs/heads/master"},"5dc7172c6c7f626760e0557efaa83fe8a0acc64f":{"kind":"REWORK","_number":10,"created":"2020-05-28 19:13:01.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/08/728108/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/10"}}},"commit":{"parents":[{"commit":"f19c97983348ff7e0ab1b6453ad321a4988ab20f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f19c97983348ff7e0ab1b6453ad321a4988ab20f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-28 19:12:33.000000000","tz":-240},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5dc7172c6c7f626760e0557efaa83fe8a0acc64f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5dc7172c6c7f626760e0557efaa83fe8a0acc64f"}]},"branch":"refs/heads/master"},"e484938285960d2e4b727f9b0561b18bf5c3b748":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-06-15 15:13:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/08/728108/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/728108/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/728108/11"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-05-14 14:52:17.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-15 15:11:44.000000000","tz":60},"subject":"[WIP] centos-8 support","message":"[WIP] centos-8 support\n\nChange-Id: Idb16322ed6554656a54214bfd4a1d5d3b92503d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e484938285960d2e4b727f9b0561b18bf5c3b748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e484938285960d2e4b727f9b0561b18bf5c3b748"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
