)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~729876","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I6b3c9f04f7a798759f67523a234f958d213c1120","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-focal","hashtags":[],"change_id":"I6b3c9f04f7a798759f67523a234f958d213c1120","subject":"Use btrfs-progs instead of btrfs-tools","status":"MERGED","created":"2020-05-21 08:45:24.000000000","updated":"2020-05-26 18:02:38.000000000","submitted":"2020-05-26 17:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729876-1590515928997-69ea3769","meta_rev_id":"9028fac6eea083dd99b469b151ca21631119c806","_number":729876,"virtual_id_number":729876,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 17:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2020-05-26 17:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-05-26 16:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2020-05-26 17:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 16:21:13.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-26 17:34:43.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-05-26 17:58:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db83870a02546e31cbcc8a7b3f90a7f9c5816e10","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 08:45:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0de480ff313bcf510f4ea586f1e21a5be47395a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 09:06:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e4d417a2a544481d8669a01e01ea0d9b : SUCCESS in 8m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9f0219b4684540a78017e91ce4e7f755 : SUCCESS in 10m 32s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/26acc331e20e4da1b4105ba6677030a4 : SUCCESS in 13m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/2f0be97b74204192b8adf3d6a5beaf9f : FAILURE in 8m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa00323e254643769a200947c6a2fb2a : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8421870d8f114309a05f624763fb70db : SUCCESS in 8m 44s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/a7caddc2144f415d89318617eb191f8c : SUCCESS in 18m 17s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d51eb9a189e84289a57846d9fe5d5548 : SUCCESS in 16m 31s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ca97c4b86a54dd3993baab7976a6f5e : SUCCESS in 15m 48s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9f903b742884be192ec1e03fbd09dbd : SUCCESS in 16m 03s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/84a69f4b9c90411fb417cc39d71c91d9 : SUCCESS in 18m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d78f7aad33e44d99544eb14fbd92c5e : SUCCESS in 15m 56s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34b2bda89a5e45528f5cfd2b6ff41dfe : SUCCESS in 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"05e3cc408e78f79cc8a240b2f30a9af3ad9aa5b6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-26 16:20:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d4a0c1a06f6974d11fefe408296fe34d7e27a883","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 16:21:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5a430baab5c658540e3ae51c44dece470449c766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:54:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/657ac10335b14f8a9000222f1cea7e0f : SUCCESS in 9m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1f195293d4864f98bda04848ba6883b8 : SUCCESS in 13m 33s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/067060d339de4105a19654cbe389473d : SUCCESS in 13m 39s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/e07d8d8e10fb430387c7bbfff4b43e11 : FAILURE in 10m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0c7a45922d845b7ad6e2b2b06ea4d0a : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ace431a9090425f8197e129971b5691 : SUCCESS in 6m 32s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/23fcf98d3eae4318b16273aae349915c : SUCCESS in 18m 55s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b3fe2747aca4a81b7649ec22f625252 : SUCCESS in 14m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8969db62c327404fb822efa293f607f5 : SUCCESS in 20m 28s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/788f5af6e14c4168b8e9b537d216beca : SUCCESS in 14m 08s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/6135afb394344fdea187a486e181da4d : SUCCESS in 20m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f55c652c47c449929e614857bd95e1bc : SUCCESS in 17m 03s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3ab9d31cf0c41fe9cbb595590c056ac : SUCCESS in 18m 44s","accounts_in_message":[],"_revision_number":2},{"id":"450632efe88b2a092ef97385a83a9e7004500617","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-05-26 17:34:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5af0b71361b32565e6825b5129ef2d7a6e4e474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:34:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af5b92cbbbd943deb7cdcf0b4cd20cc3ca4c0315","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:58:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"94f287a0ed9efcc02e0bd796f279856955d61a58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:58:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6fd8a39a99cf405b8674c69e5fb37edf : SUCCESS in 11m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/23d2d31f738c4fe5b4498d0bfe23ba60 : SUCCESS in 14m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c67cd98ea49543cca90b4f064f06ac84 : SUCCESS in 19m 09s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dca68aadc00748a7b651f22e875efb40 : SUCCESS in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e081fdfaab3d4a299dd91bc3e3f5c901 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":2},{"id":"11dee25443dc9ee12c25e99951961c89a84fbd40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:02:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb39c80298d940c7929c8713e1dbdbc9 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ed33fa0c0f0b18ed5ccd67b7455cb9dcfd85f1a3","revisions":{"c3c05e7c5ebfa7e274dd2681d707588f353334c1":{"kind":"REWORK","_number":1,"created":"2020-05-21 08:45:24.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/76/729876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/76/729876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/729876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/729876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/729876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/729876/1"}}},"commit":{"parents":[{"commit":"4b34598e359217540d72ee0c46eba18d6aaf48d7","subject":"Update docstheme for style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4b34598e359217540d72ee0c46eba18d6aaf48d7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 08:44:51.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 08:45:21.000000000","tz":60},"subject":"Use btrfs-progs instead of btrfs-tools","message":"Use btrfs-progs instead of btrfs-tools\n\nOn ubuntu focal the metapackage btrfs-tools is no longer present.\nbtrfs-progs is availabale on bionic, focal and buster.\n\nChange-Id: I6b3c9f04f7a798759f67523a234f958d213c1120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c3c05e7c5ebfa7e274dd2681d707588f353334c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c3c05e7c5ebfa7e274dd2681d707588f353334c1"}]},"branch":"refs/heads/master"},"ed33fa0c0f0b18ed5ccd67b7455cb9dcfd85f1a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-26 16:20:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/76/729876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/76/729876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/729876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/729876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/729876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/729876/2"}}},"commit":{"parents":[{"commit":"deb667df145a5504185ef532a0f0fd4704c60b16","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/deb667df145a5504185ef532a0f0fd4704c60b16"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 08:44:51.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-26 16:20:57.000000000","tz":0},"subject":"Use btrfs-progs instead of btrfs-tools","message":"Use btrfs-progs instead of btrfs-tools\n\nOn ubuntu focal the metapackage btrfs-tools is no longer present.\nbtrfs-progs is availabale on bionic, focal and buster.\n\nChange-Id: I6b3c9f04f7a798759f67523a234f958d213c1120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ed33fa0c0f0b18ed5ccd67b7455cb9dcfd85f1a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ed33fa0c0f0b18ed5ccd67b7455cb9dcfd85f1a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
