)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~729883","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I9ec6e10601c40b41fd33a241094bb188f3733f0d","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-focal","hashtags":[],"change_id":"I9ec6e10601c40b41fd33a241094bb188f3733f0d","subject":"Do not install python2 packages on systems that should be python3 only","status":"ABANDONED","created":"2020-05-21 09:11:31.000000000","updated":"2020-05-27 22:09:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5ee8e016e8a2d27ec83bd91433956e4d13042a3c","_number":729883,"virtual_id_number":729883,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-21 09:35:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 09:35:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f6776c7e46dd6cef5899e6c0738806faa10e398","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 09:11:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"713f4cea6c9aceacf417ddae2153571abf0d3d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 09:35:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/471b343ec4ff4dc79e57ddab0ad265c3 : SUCCESS in 6m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/126e2efe7f684374aaa8e64695e8f41d : SUCCESS in 11m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9054131a5add40909cde0618fcbe8f71 : SUCCESS in 13m 11s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/37a459e2ca7c4a5395ffaee374a40d83 : FAILURE in 8m 53s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b3539db226d4994ba0ef8b1c5e87205 : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49a9a2b2c514d8796a0abf50b3748f6 : SUCCESS in 9m 03s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/987eba193c4c49a1b2e8febe392fb4bb : SUCCESS in 17m 49s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5f56ac088104555a83b80d4e0a58285 : FAILURE in 12m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2932f03571fa42e2a4fef9f2c1b7787e : FAILURE in 12m 34s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c179d22a9da4483aabf25c12a2bd312 : FAILURE in 12m 13s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/7e2f73229277404c962c64d189f78301 : SUCCESS in 18m 59s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a29263c4512246e891b15732cc65823e : FAILURE in 11m 57s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1333d05a38934e099fd18c6fd6e1683b : FAILURE in 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6e5aaf128d0c1d343eb556372e487161df856ed5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 22:09:04.000000000","message":"Abandoned\n\nNow covered elsewhere","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0495044213fad372130fde8a80475baed800ec2a","revisions":{"0495044213fad372130fde8a80475baed800ec2a":{"kind":"REWORK","_number":1,"created":"2020-05-21 09:11:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/83/729883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/83/729883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/729883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/729883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/729883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/729883/1"}}},"commit":{"parents":[{"commit":"c3c05e7c5ebfa7e274dd2681d707588f353334c1","subject":"Use btrfs-progs instead of btrfs-tools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c3c05e7c5ebfa7e274dd2681d707588f353334c1"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 09:10:20.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 09:10:20.000000000","tz":60},"subject":"Do not install python2 packages on systems that should be python3 only","message":"Do not install python2 packages on systems that should be python3 only\n\nChange-Id: I9ec6e10601c40b41fd33a241094bb188f3733f0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0495044213fad372130fde8a80475baed800ec2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0495044213fad372130fde8a80475baed800ec2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
