)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~729919","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I07d5f33abd0c87aadf0da0898459eb833957ae71","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-focal","hashtags":[],"change_id":"I07d5f33abd0c87aadf0da0898459eb833957ae71","subject":"Add legacy config key mappings for lxc version 4","status":"MERGED","created":"2020-05-21 11:24:32.000000000","updated":"2020-05-26 18:00:46.000000000","submitted":"2020-05-26 17:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729919-1590515927256-231207c7","meta_rev_id":"15a6cdce960ea09c4fb6931c68ebdaaa9a32dee0","_number":729919,"virtual_id_number":729919,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 17:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2020-05-26 17:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-05-26 17:09:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2020-05-26 17:34:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 17:09:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-26 17:34:25.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-05-26 17:58:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6dfeb74d6d788fd8d5a0888a1014f8aed15084ce","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 11:24:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1795ac0fa515dc60e4b675db14364f559fdbdb81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 11:48:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0f0abd12a53d4fb994b8793fc3eef0f5 : SUCCESS in 10m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cbb3a7c925314c2fbdaee50fc7915339 : SUCCESS in 9m 05s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2f9750c9f08749e397a5ab862d8e943f : SUCCESS in 12m 13s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/9bc356c49e274539851387b3007bc772 : FAILURE in 8m 05s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2d1abcb86e8432c83fea453330deb3d : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8f7e82b5784c579abb34425a458fde : SUCCESS in 7m 03s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/15a96d34be344537ae6096ff65a41854 : SUCCESS in 17m 14s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaf17e614da442fc9b29b8fe9465a194 : SUCCESS in 15m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/946dbba6dc194538a8f365f8eb7a3229 : SUCCESS in 16m 46s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd7745956d0d4594bfbb6d1774420dac : SUCCESS in 12m 17s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/bccac624b4ec4a239d7edf389f63744f : SUCCESS in 17m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ff2defbb38748069c315d50c101a879 : SUCCESS in 15m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8ba9d4e41da465a921fd280676f6d64 : SUCCESS in 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ec7bf682cb1f7ccd021b5c69547bb52d214565ec","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 12:12:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0ed32c3b87d3b13054774bb63c72dcc7176531fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 12:39:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5386de9249344190984a9c62d40b1603 : SUCCESS in 8m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/39d4e76c1d394d70b026b697fac960cf : SUCCESS in 9m 03s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/dae2b62e2980433b835935a4ae7f2cb5 : SUCCESS in 14m 41s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4ba257662aa04a848c99b5b290aa8ad6 : FAILURE in 8m 26s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1ef17e7cc804c7189860d9b96a247f3 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a999fdd244b943799a1d8cd259f0ac8f : SUCCESS in 6m 16s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/6839e1632e7545f8af8a14c11862ebae : SUCCESS in 21m 34s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c72c1b2dc6e42ada88a26d9ec6b9cd7 : FAILURE in 18m 16s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a67f38236a774bb1852cd51b6b8ca656 : FAILURE in 13m 32s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b7c31da4819415984246a42fc3d4e94 : FAILURE in 11m 41s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/86edf0b0d26c420a923317993ab7371d : SUCCESS in 18m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a170b52e880643c2a90cccf387c96f06 : FAILURE in 11m 18s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3965ae0b31d46258b006899ca8d4a21 : FAILURE in 14m 22s","accounts_in_message":[],"_revision_number":2},{"id":"12d419235a8fb7c78d4409fb581a2513744ece8d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 19:12:16.000000000","message":"Topic set to osa-focal","accounts_in_message":[],"_revision_number":2},{"id":"2b3dea9a151f00f562d9caa9e485567bb9cc470a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-26 16:20:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0d0439d3ca69027cd3e7dc1e828767bfee14bbb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:56:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4623c82c61434712b2c1a1198a94a171 : SUCCESS in 9m 07s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8173faaeba9b482e8acc88375f12d068 : SUCCESS in 9m 37s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/710e67eff3b6492aa6903531459fcac4 : SUCCESS in 21m 40s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c150f78870aa4627996d24c8da16ef49 : FAILURE in 11m 36s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b548a827d064c2b8c8885532e791135 : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0487d9776574f69b10c65cdbce9ec23 : SUCCESS in 9m 19s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/73c9dfc217894ef88ca80b12fb9483f5 : SUCCESS in 24m 44s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/797bba8a265148ae82f7bf27fd188769 : SUCCESS in 18m 49s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4401706b75a9461c93e6a24c1c2f46a5 : SUCCESS in 20m 41s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db3ff0408d044fee90d7091c81ea7b68 : SUCCESS in 20m 25s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/a889dab2a54a4a41a4e4f5e6c29dd768 : SUCCESS in 29m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e166de2a20d4df1aa67736e3225d49a : SUCCESS in 15m 30s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a4576e70cd148aea727df3eff67d4f2 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1c53966d92f157799caa716d723d4fae9e818f1e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 17:09:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"225de10073e84a42fb008d51665dc7e6cf561929","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-05-26 17:34:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"684bf96f9e36c4a50af06faf5656c29ea563bcfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:34:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"54ec975d14c38e5ffffcda8f28dc723e206ea8e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:58:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/34815a3d00aa4cd68eeb2034104f0a9e : SUCCESS in 10m 17s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ed9dc0f8ddd744ae8ff83cc4f07e9402 : SUCCESS in 11m 36s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b29efd35682a4215953ce804b10d0f78 : SUCCESS in 20m 51s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/888fbc7b06ed40c9ae6e81c5d6ca52d8 : SUCCESS in 16m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b5f45c395b4e17bab4f1899df52f5a : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":3},{"id":"6b12e0606db3b6572922d758b271abf787454484","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:58:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"15a6cdce960ea09c4fb6931c68ebdaaa9a32dee0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:00:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f058c4c38af4b90ae3db6886d624b50 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ec5163b1df89bd9ac8c26403c21709602689877a","revisions":{"8d020a3589f4c230cff39d669b717f595c4f7086":{"kind":"REWORK","_number":1,"created":"2020-05-21 11:24:32.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/19/729919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/19/729919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/1"}}},"commit":{"parents":[{"commit":"d3c1910cdcb37b2930149860d2fff7e67901edd2","subject":"Add py3 installation for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d3c1910cdcb37b2930149860d2fff7e67901edd2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 11:23:57.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 11:23:57.000000000","tz":60},"subject":"Add legacy config key mappings for lxc version 4","message":"Add legacy config key mappings for lxc version 4\n\nChange-Id: I07d5f33abd0c87aadf0da0898459eb833957ae71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8d020a3589f4c230cff39d669b717f595c4f7086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8d020a3589f4c230cff39d669b717f595c4f7086"}]},"branch":"refs/heads/master"},"7024892bb7c510fc4c73093ebc66c46555020ddf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-21 12:12:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/19/729919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/19/729919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/2"}}},"commit":{"parents":[{"commit":"0495044213fad372130fde8a80475baed800ec2a","subject":"Do not install python2 packages on systems that should be python3 only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0495044213fad372130fde8a80475baed800ec2a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 11:23:57.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 12:12:19.000000000","tz":0},"subject":"Add legacy config key mappings for lxc version 4","message":"Add legacy config key mappings for lxc version 4\n\nChange-Id: I07d5f33abd0c87aadf0da0898459eb833957ae71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7024892bb7c510fc4c73093ebc66c46555020ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7024892bb7c510fc4c73093ebc66c46555020ddf"}]},"branch":"refs/heads/master"},"ec5163b1df89bd9ac8c26403c21709602689877a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-26 16:20:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/19/729919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/19/729919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/729919/3"}}},"commit":{"parents":[{"commit":"deb667df145a5504185ef532a0f0fd4704c60b16","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/deb667df145a5504185ef532a0f0fd4704c60b16"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 11:23:57.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-26 16:20:43.000000000","tz":0},"subject":"Add legacy config key mappings for lxc version 4","message":"Add legacy config key mappings for lxc version 4\n\nChange-Id: I07d5f33abd0c87aadf0da0898459eb833957ae71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ec5163b1df89bd9ac8c26403c21709602689877a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ec5163b1df89bd9ac8c26403c21709602689877a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
