)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~729937","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I2d0d42046d4535113deb75f8b22e7c33489b6f5f","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-focal","hashtags":[],"change_id":"I2d0d42046d4535113deb75f8b22e7c33489b6f5f","subject":"Add vars files and container prep script for ubuntu focal","status":"MERGED","created":"2020-05-21 12:15:58.000000000","updated":"2020-05-26 17:58:48.000000000","submitted":"2020-05-26 17:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729937-1590515928116-231207c7","meta_rev_id":"48b4ea68b25ae784307e636c3a35698b84b9b619","_number":729937,"virtual_id_number":729937,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 17:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2020-05-26 17:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-05-26 17:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2020-05-26 17:31:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 17:08:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-26 17:31:55.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-05-26 17:58:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f41f7cb568dc75e9e0b7b807d835ac324f479940","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 12:15:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dacf1d3ce79f5e1685bd5c4417adfebc1a6d3692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 12:48:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e9d8266275334919ae6b973e80943a2d : SUCCESS in 11m 04s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f0391bf1301641d39a57d8654ee7b131 : SUCCESS in 11m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/90e6f7fa0131426b83bf40b21d56f83b : SUCCESS in 17m 40s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/dc1337b719834ea1907814cd92fdcc48 : FAILURE in 8m 38s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4200b1740be74aecbf9fde8eb853b90f : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/471742d314354f5aa0d3494eda7220fd : SUCCESS in 6m 58s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/44eb6ee4a84d420eb7134b74c1ecc6e4 : SUCCESS in 17m 40s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aad2a507575547a1a625fc2de7c8f083 : FAILURE in 15m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a65a44ec82b94bd799afab789b156e8b : FAILURE in 17m 28s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a075ca4b9884bd582ea76f289fa5ba3 : FAILURE in 12m 33s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/1f0d4562a0ae44f7955fad3648f22fa6 : SUCCESS in 29m 21s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65d4e7601f8b400fabd9a9509b4af4e3 : FAILURE in 14m 27s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fb35f063b364b2fa9891d4f8db48f8d : FAILURE in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f55536660d2f7fdad1fb60096a210610b38dea9b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 14:11:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d790f09297a7e2893315ec745bc76a3607417ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 14:39:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7f32dddc28a54fdb979a07236ff519ad : SUCCESS in 8m 22s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b2e42bcaf3ca4169b6ae706f8a59af11 : SUCCESS in 11m 39s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3a29ff5fd7354148a485a30a0bc6f0fb : SUCCESS in 12m 57s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/da0bae0932a346b7afb8d0a54b0d2000 : FAILURE in 10m 13s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02821eb8473f47dd9761ce0bd2928765 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c7ca3094d746e6be4b324ea8990867 : SUCCESS in 8m 57s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/3b15279c5eab4e68840d698bf297aca1 : SUCCESS in 22m 47s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b30e2c18e65a4388a8f74bd892e1e207 : FAILURE in 15m 42s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/830c5dfbd0654cfd8cf886e5ec6d305f : FAILURE in 15m 38s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e781133f5634cbc9611b035a27c2a9b : FAILURE in 14m 51s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/6c3be516f933425aac5bc956a8cbaa85 : SUCCESS in 17m 07s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f03da3302d82460cbc757689f9bc791e : FAILURE in 14m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11fe66e669a84fa1a62df5b2d5e317b8 : FAILURE in 14m 27s","accounts_in_message":[],"_revision_number":2},{"id":"694ad720325c6fe602a3da62b5dc503dbac594d5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 19:12:02.000000000","message":"Topic set to osa-focal","accounts_in_message":[],"_revision_number":2},{"id":"ce0ef7b301d8eec8d8b9fa295fcedde486c96f8e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-26 16:21:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bc1b4aa1ee1e5042d091d0725131028872bff27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:57:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/98609ab071c942d6901b2757adf67fc9 : SUCCESS in 9m 46s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9c6c18850e5c442bb9772fd1ecb8b5b9 : SUCCESS in 10m 11s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d4732b809493488380471dd0ae8684ae : SUCCESS in 11m 02s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/e02e4058603f46eeaf86c90123b5d53d : FAILURE in 11m 08s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6692eb89d554dec8c868f7510f8818d : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55e139af93647239558b115c2ad1113 : SUCCESS in 7m 20s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/cd8a525a5dc14501b4af7a7ac6f4a048 : SUCCESS in 19m 52s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22ab22c4b7f84e22846473b91543e033 : SUCCESS in 17m 31s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/075841967966413d94dfec0e91b35560 : SUCCESS in 20m 38s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/598c70f69bbe4b5e969e0804f07af5f4 : SUCCESS in 13m 54s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/76aded6f598943389dce3e7a4fb082b5 : SUCCESS in 17m 43s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cff7d726aab463693fb71bbf998a249 : SUCCESS in 16m 15s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e62394887e244ccae03ad0b4bb2399b : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":3},{"id":"cea0cd06b1c1f9b2a8e8dfcec0a39ba95b551e07","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 17:08:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"53d911073eaf3b0f13720da1ce4c995e6f95df06","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-05-26 17:31:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"884ecf2b43cb4bf3d5f3a86477c28987262ef7cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:34:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d6b009e39d042420346cad4be770fb11cd105ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:58:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e890512fc78b4053aa0e9ea356ef513d : SUCCESS in 7m 43s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fc815268bc664ef480800325ff6ed3f0 : SUCCESS in 11m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/11c77bc54713445195b16aa9deddc374 : SUCCESS in 12m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa86c6fe41bc4c189470820ea307b594 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7c973a662546139ce4225a5349298b : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":3},{"id":"48b4ea68b25ae784307e636c3a35698b84b9b619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:58:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4d16a8883ead99635c405e76d5c4b648151a0a3d","revisions":{"04a720ba65593f5856fefd8c2e4af6b3570d2a44":{"kind":"REWORK","_number":1,"created":"2020-05-21 12:15:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/729937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/729937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/1"}}},"commit":{"parents":[{"commit":"7024892bb7c510fc4c73093ebc66c46555020ddf","subject":"Add legacy config key mappings for lxc version 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7024892bb7c510fc4c73093ebc66c46555020ddf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 12:15:30.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 12:15:30.000000000","tz":60},"subject":"Add vars files and container prep script for ubuntu focal","message":"Add vars files and container prep script for ubuntu focal\n\nChange-Id: I2d0d42046d4535113deb75f8b22e7c33489b6f5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/04a720ba65593f5856fefd8c2e4af6b3570d2a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/04a720ba65593f5856fefd8c2e4af6b3570d2a44"}]},"branch":"refs/heads/master"},"b3b25433c80e977d2b476fe6d452b2381b01ba0c":{"kind":"REWORK","_number":2,"created":"2020-05-21 14:11:27.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/729937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/729937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/2"}}},"commit":{"parents":[{"commit":"7024892bb7c510fc4c73093ebc66c46555020ddf","subject":"Add legacy config key mappings for lxc version 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7024892bb7c510fc4c73093ebc66c46555020ddf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 12:15:30.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 14:11:18.000000000","tz":0},"subject":"Add vars files and container prep script for ubuntu focal","message":"Add vars files and container prep script for ubuntu focal\n\nChange-Id: I2d0d42046d4535113deb75f8b22e7c33489b6f5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b3b25433c80e977d2b476fe6d452b2381b01ba0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b3b25433c80e977d2b476fe6d452b2381b01ba0c"}]},"branch":"refs/heads/master"},"4d16a8883ead99635c405e76d5c4b648151a0a3d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-26 16:21:32.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/729937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/729937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/729937/3"}}},"commit":{"parents":[{"commit":"ec5163b1df89bd9ac8c26403c21709602689877a","subject":"Add legacy config key mappings for lxc version 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ec5163b1df89bd9ac8c26403c21709602689877a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-21 12:15:30.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-26 16:21:32.000000000","tz":0},"subject":"Add vars files and container prep script for ubuntu focal","message":"Add vars files and container prep script for ubuntu focal\n\nChange-Id: I2d0d42046d4535113deb75f8b22e7c33489b6f5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4d16a8883ead99635c405e76d5c4b648151a0a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4d16a8883ead99635c405e76d5c4b648151a0a3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
