)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~731177","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ib914223ee30c5f2780f04972e013f3872195a79d","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-focal","hashtags":[],"change_id":"Ib914223ee30c5f2780f04972e013f3872195a79d","subject":"Fix systemd version detection for ubuntu focal","status":"MERGED","created":"2020-05-27 11:59:52.000000000","updated":"2020-05-27 15:26:37.000000000","submitted":"2020-05-27 15:24:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731177-1590593053407-65991141","meta_rev_id":"3c466d42dfb9b5bef31e61db415395a40eceba92","_number":731177,"virtual_id_number":731177,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 15:24:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2020-05-27 13:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":2,"date":"2020-05-27 15:05:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":1,"date":"2020-05-27 15:05:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 13:08:45.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-05-27 15:05:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-27 15:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e901d389f58cc1cbeecc007217f28af7065c90b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 11:59:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ecccfd90519d311113d0aab847b04cd55fe00b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 12:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2b94e0bcb7744a2f828c6778c5f3dd63 : SUCCESS in 11m 34s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f897731292114295b03ce0572d24ecca : SUCCESS in 12m 08s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ac6494115a364fbdb0729ceaa6987332 : SUCCESS in 19m 43s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/350f9fba2280464fbd79c10ed4f93557 : FAILURE in 8m 18s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbc84f522610473cbbd6e6da6f8a2255 : SUCCESS in 11m 51s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9afbf94b2834830acd60e2b8bf45c6f : SUCCESS in 14m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a43790121041f68cb7a33c2717ec0e : SUCCESS in 5m 52s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/30cd2c6a17be43dd8bea24cd5b770098 : SUCCESS in 19m 54s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98531723df2f448e98a9a7b19d1f0dde : SUCCESS in 14m 25s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfd9a5f9d59c453499c62aa6e826bacf : SUCCESS in 13m 21s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff25db60024b4b26a2df756e2cfcd34b : SUCCESS in 14m 41s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/a98fa4ee080543f38ca64c2bbafae6c7 : SUCCESS in 20m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b031d88bdb54434dbde08403ae619f73 : SUCCESS in 17m 01s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d60dc6958eec4a1789a1c36fdd0b9768 : SUCCESS in 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9fa934c7bd3e01d21f423425f15e75ea221f9941","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 12:33:57.000000000","message":"Topic set to osa-focal","accounts_in_message":[],"_revision_number":1},{"id":"25ab540398f1cfe3d6d5767a4ec21fc481b5ff1b","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-05-27 13:08:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5631ac3ecfa6189dc26afe87e3e486a4986812b9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-27 15:05:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f06d7aaa1eec5a6daf1bed897f2c7f9ae217842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:05:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88f0b024bf1195fd205323b7c2decbd784bc2468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:24:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ca47b31be8b44d40b7321e7d3ac0897d : SUCCESS in 9m 03s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/683449f112bd494494450c860f4c6c20 : SUCCESS in 10m 22s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/fe60a76761d247f4998bfba0d398c06a : SUCCESS in 11m 01s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/761eb60b129d4e599b3b3067006135cc : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14f837c77b94694a56223a5deb623e5 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a3f14358acac7006af44481d0dd4137188e272b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:24:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3c466d42dfb9b5bef31e61db415395a40eceba92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:26:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d68fc1d3e98248e89591f753c05027a9 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6699442bad4f5391b9086905c15974284af11edd","revisions":{"6699442bad4f5391b9086905c15974284af11edd":{"kind":"REWORK","_number":1,"created":"2020-05-27 11:59:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/77/731177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/77/731177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/77/731177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/77/731177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/77/731177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/77/731177/1"}}},"commit":{"parents":[{"commit":"c8abe593118f6971d69d7ab71019eb9a503eb637","subject":"Merge \"Use btrfs-progs instead of btrfs-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c8abe593118f6971d69d7ab71019eb9a503eb637"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 10:37:29.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 10:37:29.000000000","tz":60},"subject":"Fix systemd version detection for ubuntu focal","message":"Fix systemd version detection for ubuntu focal\n\nOutput from \u0027systemd --version\u0027\n\nbionic: systemd 237\nfocal: systemd 245 (245.4-4ubuntu3.1)\n\nTo make this work for either format we should split the line and take\nthe second element, rather than the last.\n\nChange-Id: Ib914223ee30c5f2780f04972e013f3872195a79d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6699442bad4f5391b9086905c15974284af11edd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6699442bad4f5391b9086905c15974284af11edd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
