)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~731178","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I6a441b4e3362f4ddccffb92cb8227e37a7a3656f","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-focal","hashtags":[],"change_id":"I6a441b4e3362f4ddccffb92cb8227e37a7a3656f","subject":"Fix ubuntu focal to use correct image","status":"MERGED","created":"2020-05-27 11:59:52.000000000","updated":"2020-05-27 15:30:27.000000000","submitted":"2020-05-27 15:24:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731178-1590593054116-c44eae83","meta_rev_id":"915530ff98c2982875c45697a11b6c2ca39d1a13","_number":731178,"virtual_id_number":731178,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 15:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2020-05-27 12:52:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":2,"date":"2020-05-27 12:56:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":1,"date":"2020-05-27 12:56:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 12:52:29.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-05-27 12:56:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-27 15:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"320caa881c0069f174bdbe6e13071cbe256fe7e6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 11:59:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abec40e9bece2cde65aa66ccce0ce8eef04221ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 12:29:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e25d7b068d5a4795bc171476c994d94d : SUCCESS in 7m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/3aece977fdf04f9d88bacc4c6c1ff06b : SUCCESS in 10m 06s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b2f7c21a309540adaa68c6f3becafc8a : SUCCESS in 14m 39s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/18d0c107fd4c404f8af4bad847420e8a : FAILURE in 9m 27s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a042bb64e13047c1b239c0c4a3840aa7 : SUCCESS in 10m 44s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c358c0a9cb0a49d9b1f60d7d94f30d0c : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e118158b284fc1ac2d3dc77a70bb92 : SUCCESS in 5m 46s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/c188d760fa914727955894d322cebb39 : SUCCESS in 24m 11s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4faaad36409a41fcb739ed3fa74a7a47 : SUCCESS in 16m 02s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b08d0719775643a1a7cd2afb34a368d1 : SUCCESS in 14m 24s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f16010b30be64f148e52f1068c9f9d90 : SUCCESS in 15m 36s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/90ece21e492e4f6695196001c9e3a291 : SUCCESS in 18m 46s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81be81776cab4fc9918a9e468d48b060 : SUCCESS in 14m 43s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a5ad5b9d1cb412fbedd70023ebd2f85 : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"14257b118d6daab42f2b293e6bdbdc8194de0a7e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 12:33:19.000000000","message":"Topic set to osa-focal","accounts_in_message":[],"_revision_number":1},{"id":"5ae42d448e5d9b5a03a1f57cc0d569a918c434c5","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-05-27 12:52:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e97e68176ab7c71e67c3d3217ee3053334052ac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-27 12:56:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f5f13deb796e4557b4fb0d1b571b69fccc763c1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:06:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c655128263398b60d5fa62085dda0a27546f86f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:24:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a59e6b6cba964450a202bf2859dce973 : SUCCESS in 8m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2e76bf2af73e419eb4281de4ab8e42c4 : SUCCESS in 9m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ebc1274b81eb4b14a139631cbab370a8 : SUCCESS in 10m 55s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6822ddb6b3054113941b8eef2407c341 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1baf3f01de1c466e9b2120691c99369e : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ba6e3aa0658bf5a9d241b84a5382fb9a433b0bca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:24:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"915530ff98c2982875c45697a11b6c2ca39d1a13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:30:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8afaa37e49714d178d7d67e9de45790e : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e25305c7585845007d17299ec0bd4583b73e41c","revisions":{"7e25305c7585845007d17299ec0bd4583b73e41c":{"kind":"REWORK","_number":1,"created":"2020-05-27 11:59:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/731178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/731178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/731178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/731178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/731178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/731178/1"}}},"commit":{"parents":[{"commit":"6699442bad4f5391b9086905c15974284af11edd","subject":"Fix systemd version detection for ubuntu focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6699442bad4f5391b9086905c15974284af11edd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 11:59:06.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 11:59:06.000000000","tz":60},"subject":"Fix ubuntu focal to use correct image","message":"Fix ubuntu focal to use correct image\n\nChange-Id: I6a441b4e3362f4ddccffb92cb8227e37a7a3656f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
