)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~731228","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ia9bc8feb385cbc7dda8ed6163ae208bca845ac0f","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-focal","hashtags":[],"change_id":"Ia9bc8feb385cbc7dda8ed6163ae208bca845ac0f","subject":"Use a single config for container prep on ubuntu","status":"ABANDONED","created":"2020-05-27 14:43:16.000000000","updated":"2023-02-01 15:39:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9dcec0b250c270085667ce96204ba4fff680463f","_number":731228,"virtual_id_number":731228,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-27 21:52:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-05-31 09:08:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 21:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4726931d105ae2b918b50207b9eef4ec2c023036","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 14:43:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ea5511a2d6520d7d766c764d28eedc8df0d7add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:16:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/42b94af8ee5d4bc9befd4d1e7e2eba89 : SUCCESS in 7m 22s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c677e3a1a9774677894c17be30c1dce6 : SUCCESS in 10m 24s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/66dd89adf92e4b59830fee4a9a6838e4 : SUCCESS in 14m 17s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/5587dbdb23e04189add8ac229df45123 : FAILURE in 9m 29s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f45442e8490e41d5bf9d505abe3da0ea : SUCCESS in 11m 47s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95fe5f79189e4026b166cb88ba88f889 : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5deae1e7937145d782faea51bcd5f266 : SUCCESS in 8m 16s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/e42a58308b3e405ba31ec5a56a514613 : SUCCESS in 19m 57s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3a8e63e1bba4bf898974a3f8aa81c12 : SUCCESS in 15m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/048f45b1b32048a6982d5aa6fc868b67 : SUCCESS in 16m 56s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da36a982ee624e32967406af240135d1 : SUCCESS in 19m 09s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/17b60af8d781456097d17187e87fbab3 : SUCCESS in 20m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2966de842a554a4b90faa28f1749d01f : SUCCESS in 16m 28s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa3413aa92f3499791391c68f526314c : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fdf3924b4879ee3bce797318240f09e74d50f8c2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 15:21:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e9b0f63265a4e18eefcfc9d6af2aa8284b6d1b7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 15:22:45.000000000","message":"Topic set to osa-focal","accounts_in_message":[],"_revision_number":2},{"id":"0b2d82e4a5bb6e21e26f0eedc7467d6217f2f059","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 15:38:54.000000000","message":"Patch Set 2: Code-Review-1\n\nshould try this without python-is-python3 to see if something sensible happens with interpreter discovery","accounts_in_message":[],"_revision_number":2},{"id":"9bcee9284f951279c67b5be730e2cd4f4b5dc522","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 15:40:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0da095e0c878bd5934272a93870f567c9bd3a8c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 16:09:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fc25da2260cd4fcba5ddf07eea8d0979 : SUCCESS in 9m 51s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c769d1c9611b4b048e1f1a659ce0c60a : SUCCESS in 10m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/edeab5fcc49e47bd985736404affd669 : SUCCESS in 15m 29s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ca6362bc443c467b95fb0b67300715fa : FAILURE in 9m 46s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3199707abc964a4b9ebf81e1417f36f3 : FAILURE in 11m 39s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae3cfcd431dd4610b88bb5eb6655028c : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68cad9b8bb4048baa8ee65f3645141d7 : SUCCESS in 9m 03s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/5dfb3bc43dab427bb8cb5a6c39d1c4b8 : SUCCESS in 18m 27s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd478ef87ccc40cf94266989becf7222 : FAILURE in 11m 29s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b3bca1064404247877a25744ff82dcd : FAILURE in 11m 37s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48a548b967044aaba255bacc7314b7d1 : FAILURE in 9m 40s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/d353d5d5c8254e5cb99a44a2ef69410f : SUCCESS in 20m 18s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/926c332b51644fcc9210dd5f690b38b6 : FAILURE in 12m 19s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ee6aeab13e74c12b4e2bb7f549b8f6e : FAILURE in 12m 47s","accounts_in_message":[],"_revision_number":3},{"id":"9775d86da184c3dfe516d8b3a242474017124422","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 17:22:10.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"75ce01c4399c9e4bfd474a11890e2eaf2858d8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 17:52:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b6a46fdc951f45ecbcb873cc03d928e5 : SUCCESS in 9m 26s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1142a4ba799c41f8aac4b769ad390b81 : SUCCESS in 10m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4e22c19313b1454f87c0607e8674e476 : SUCCESS in 12m 18s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/6d90c903ba104ef09c7d3515da8e2796 : FAILURE in 7m 02s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abc2611ea1724e05bffa6633a1ba7358 : FAILURE in 7m 34s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/739a300816694814bb2aa388bc2c0b49 : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aeb00f53803463aaa7f65c584deed53 : SUCCESS in 6m 19s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/f4276597d9524effb2cba70f5f3701d0 : SUCCESS in 17m 01s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9166737b389944098113344c075c03b0 : FAILURE in 11m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54c003684dac4936a3d96dfd07cb24dd : FAILURE in 11m 28s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db70be7afcc74eed80b87d2a6463db8c : FAILURE in 12m 15s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/bf63dc9511eb4e1599c071537b507c91 : SUCCESS in 18m 03s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34c381ea8c904a888a1c642d97a9ba10 : FAILURE in 12m 08s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21a6c98a2b78452ea6743aa67b11a400 : FAILURE in 11m 05s","accounts_in_message":[],"_revision_number":4},{"id":"9964e09d56529cfe3825e06c2d875904bca2901d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 20:39:14.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2e6c39304d5416f88da6902ce94eeeff29b8a424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:10:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/838d3b561a4f43e2b5cfc5087fc43fcc : SUCCESS in 8m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/65c4c0e00a9a495ca0ccc5c3a85605df : SUCCESS in 9m 29s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a001b574fa3247eabaec1179b4c982d1 : SUCCESS in 15m 01s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/33c02a7eacfb40db98bf42e30484327b : FAILURE in 8m 14s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/892d41998e3c42198461a8f41e25b239 : FAILURE in 10m 37s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d67ea98cfbb4079985b700c16659efd : FAILURE in 10m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651229effb2045ddaf524e5b1c0ac17d : SUCCESS in 8m 54s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/b2c1dbdf3a334067986a0d04dd44b148 : SUCCESS in 29m 02s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcc5e0e151ad453fb3446811cdb64591 : FAILURE in 10m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfe0e469260d481a81cff76ada95a1f2 : FAILURE in 12m 10s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da9851c958204c5380530e46504b7a3b : FAILURE in 10m 16s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/8d530b63aeed4e688a7b3b00b0037fca : SUCCESS in 18m 02s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07310ed377c642488b2ff9dfedbf5443 : FAILURE in 11m 54s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/984068a66b1f4101bbee57f7086d7c2e : FAILURE in 12m 19s","accounts_in_message":[],"_revision_number":5},{"id":"594b60b6f158398415dfb24a6a8790bbf73e0fb6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 21:17:20.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a1135f59d8e3fb38ea1fdb51aa1082d2398f1823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:52:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2568e5bb9b174c81a335b03f13a241de : SUCCESS in 10m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d5d22eb7e136426ba9b49e21ebb0e62c : SUCCESS in 9m 42s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c6aff9af26c94bc1bba572d04b171057 : SUCCESS in 13m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8444580e2ce64eb49024e883461bfb18 : FAILURE in 12m 09s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/628ce7a8307b44daa691345e4b4f6d02 : SUCCESS in 10m 56s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/805e1e24892d49d1bdc6455cccfdf1a5 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37828fad8a184063a1768a428d760176 : SUCCESS in 6m 32s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/79ed6a03d77b4977977168260b289269 : SUCCESS in 19m 25s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bec6f5cd3a14eec9714021a39c44803 : SUCCESS in 17m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eac19663a9944b1ead371026064ef477 : SUCCESS in 18m 10s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae90042a4449447f87d1c605cf2d418c : SUCCESS in 16m 05s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/cc064b08096b4f9d99d194f50d09bea1 : SUCCESS in 19m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/719fd200337e4202bd8bc3c289e4b0bc : SUCCESS in 20m 33s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0d3cca316e849f885f368e06c1bb244 : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":6},{"id":"f54cf1b1f09f7839d61cdd6b4370d369a3a2229c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-28 13:55:52.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"cb89877141fee9d2bb04735648c3e48faa3c1a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:28:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1c3fb6413a3549d28532ac1af52c6970 : SUCCESS in 7m 25s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/019d20c4fb474c278c9918fe96fc442f : SUCCESS in 13m 25s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f23256a8fcfc4a1b9c4aab2c9509429e : SUCCESS in 9m 26s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/e30594ec940f4e19ad2a07e80c19289b : FAILURE in 6m 54s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ac5cf51231f4a89bb46ea6d15f8b13c : SUCCESS in 9m 15s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8043d268d93b4242ba9fec866496c519 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd76b4f6fc745c6a84af14d0c7745f1 : SUCCESS in 6m 16s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/92f3dfb93788407fa245ca1abbdd8f90 : SUCCESS in 19m 38s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c8395befabd4489ae65949da3072399 : SUCCESS in 16m 51s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22257951f55c4de58e35ad826f3ec082 : SUCCESS in 14m 45s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04dd6964ecac4fc7aeb9fd18086f98fd : SUCCESS in 14m 30s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/53a5ca8226bb4c6098e2352a2f75c25b : SUCCESS in 18m 02s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ced5c7d07958484aa9fbf98e0879728c : SUCCESS in 14m 10s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6959c4c0570b4da2bda47df1d8ae8a2a : SUCCESS in 18m 30s","accounts_in_message":[],"_revision_number":6},{"id":"9f1fe5a85077c257bb1ceb8b19f878efcb90df91","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-31 09:08:32.000000000","message":"Patch Set 6: Workflow-1\n\nInfra CI image has python2 at /usr/bin/python which means we can\u0027t do this until dropping bionic.","accounts_in_message":[],"_revision_number":6},{"id":"9dcec0b250c270085667ce96204ba4fff680463f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-02-01 15:39:25.000000000","message":"Abandoned\n\nHugely out of date now","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9578d88ebccec16dfef6e9fb752e589fc071177c","revisions":{"49a03e27983930722a1735adcef8eefe6b8e8582":{"kind":"REWORK","_number":1,"created":"2020-05-27 14:43:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/28/731228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/731228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/1"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 14:11:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 14:11:59.000000000","tz":60},"subject":"Ensure python3 installation inside container matches focal host","message":"Ensure python3 installation inside container matches focal host\n\nUpdate the prep script which was copied from bionic to be\nappropraite for focal.\n\nChange-Id: Ia9bc8feb385cbc7dda8ed6163ae208bca845ac0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/49a03e27983930722a1735adcef8eefe6b8e8582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/49a03e27983930722a1735adcef8eefe6b8e8582"}]},"branch":"refs/heads/master"},"0ec9f646ba0b5e55b516de55271bd0aefe65e2a8":{"kind":"REWORK","_number":2,"created":"2020-05-27 15:21:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/28/731228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/731228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/2"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 14:11:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 15:21:22.000000000","tz":60},"subject":"Use a single config for container prep on ubuntu","message":"Use a single config for container prep on ubuntu\n\nDrop support for xenial, and combine bionic and focal into the\nsame set of prep templates and ansible vars.\n\nEnsure that the python 3 installation inside the container matches\nthe host. Do not install python2 into the container.\n\nChange-Id: Ia9bc8feb385cbc7dda8ed6163ae208bca845ac0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0ec9f646ba0b5e55b516de55271bd0aefe65e2a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0ec9f646ba0b5e55b516de55271bd0aefe65e2a8"}]},"branch":"refs/heads/master"},"9180cdc74a302a843ce7708100b9ee406482279a":{"kind":"REWORK","_number":3,"created":"2020-05-27 15:40:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/28/731228/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/731228/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/3"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 14:11:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 15:40:21.000000000","tz":60},"subject":"Use a single config for container prep on ubuntu","message":"Use a single config for container prep on ubuntu\n\nDrop support for xenial, and combine bionic and focal into the\nsame set of prep templates and ansible vars.\n\nEnsure that the python 3 installation inside the container matches\nthe host. Do not install python2 into the container.\n\nChange-Id: Ia9bc8feb385cbc7dda8ed6163ae208bca845ac0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9180cdc74a302a843ce7708100b9ee406482279a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9180cdc74a302a843ce7708100b9ee406482279a"}]},"branch":"refs/heads/master"},"19125f07f6c3c80d4d1ead4e11f14c1d3adbd6cf":{"kind":"REWORK","_number":4,"created":"2020-05-27 17:22:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/28/731228/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/731228/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/4"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 14:11:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 17:22:07.000000000","tz":0},"subject":"Use a single config for container prep on ubuntu","message":"Use a single config for container prep on ubuntu\n\nDrop support for xenial, and combine bionic and focal into the\nsame set of prep templates and ansible vars.\n\nEnsure that the python 3 installation inside the container matches\nthe host. Do not install python2 into the container.\n\nChange-Id: Ia9bc8feb385cbc7dda8ed6163ae208bca845ac0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/19125f07f6c3c80d4d1ead4e11f14c1d3adbd6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/19125f07f6c3c80d4d1ead4e11f14c1d3adbd6cf"}]},"branch":"refs/heads/master"},"2d39fac07aea22453a5deeaaed5a9fc0b971544d":{"kind":"REWORK","_number":5,"created":"2020-05-27 20:39:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/28/731228/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/731228/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/5"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 14:11:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 20:39:08.000000000","tz":0},"subject":"Use a single config for container prep on ubuntu","message":"Use a single config for container prep on ubuntu\n\nDrop support for xenial, and combine bionic and focal into the\nsame set of prep templates and ansible vars.\n\nEnsure that the python 3 installation inside the container matches\nthe host. Do not install python2 into the container.\n\nChange-Id: Ia9bc8feb385cbc7dda8ed6163ae208bca845ac0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2d39fac07aea22453a5deeaaed5a9fc0b971544d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2d39fac07aea22453a5deeaaed5a9fc0b971544d"}]},"branch":"refs/heads/master"},"9578d88ebccec16dfef6e9fb752e589fc071177c":{"kind":"REWORK","_number":6,"created":"2020-05-27 21:17:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/28/731228/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/731228/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731228/6"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 14:11:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 21:17:18.000000000","tz":0},"subject":"Use a single config for container prep on ubuntu","message":"Use a single config for container prep on ubuntu\n\nDrop support for xenial, and combine bionic and focal into the\nsame set of prep templates and ansible vars.\n\nEnsure that the python 3 installation inside the container matches\nthe host. Do not install python2 into the container.\n\nChange-Id: Ia9bc8feb385cbc7dda8ed6163ae208bca845ac0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9578d88ebccec16dfef6e9fb752e589fc071177c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9578d88ebccec16dfef6e9fb752e589fc071177c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
