)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~731625","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Ftrain~I4b74670edd7b4ae9710dbb39deb82c0775a1ff66","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/train","topic":"disable-listchanges","hashtags":[],"change_id":"I4b74670edd7b4ae9710dbb39deb82c0775a1ff66","subject":"ensure 20listchanges is not in debian container","status":"MERGED","created":"2020-05-28 23:30:00.000000000","updated":"2021-10-08 10:29:46.000000000","submitted":"2021-10-08 10:27:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731625-disable-listchanges","meta_rev_id":"bd014050ad2f7abee77b75cbed2e66f446a8b8f9","_number":731625,"virtual_id_number":731625,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2021-10-05 16:56:54.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-08 10:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-01-18 19:41:02.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-10-08 10:00:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 19:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-10-08 10:00:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 00:04:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-26 17:11:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-10-08 10:00:20.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b8aac791cf01df1ef21c74c04b16ebef49c1f313","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 23:30:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4edc2edb0713d53b601d48acb2e9ee711f41a9f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 23:43:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"069ee63a987eb13ba962b31776bc53928671c39d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 00:04:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0274b639bf664a9d9ed71b59055706d2 : SUCCESS in 6m 21s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/576bca9e9ca54dbb8c14de2420cf50a0 : FAILURE in 4m 09s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/42bcf55066ed4007be9e044de0b87854 : FAILURE in 5m 15s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a413b2d59f94e0ea8d885b281eb2d70 : FAILURE in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e75cb778e5040fcab62c4cf6b357290 : SUCCESS in 5m 14s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/51492f723e1d4ca48c229b8054d57fca : SUCCESS in 15m 53s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/481cc498f49c476aa6f1d2a2b2b52cd0 : FAILURE in 8m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c37761622c16429e9d4151559c65f22f : FAILURE in 13m 01s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fced54a5913d4a07b0f08f6136517509 : FAILURE in 8m 23s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/03d60342907d42ca87f3a2d1b1ad8b26 : SUCCESS in 15m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a943b51c597406db27b96870a0d01f2 : FAILURE in 8m 14s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/813cb97c45fc434b866589391a1b54e1 : FAILURE in 8m 34s","accounts_in_message":[],"_revision_number":2},{"id":"1f14d6ea76e4b8ae2a5aec1b51f261cf1ed01825","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 19:58:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"160a496451c1f8c1bf240bf59e22d3b1c116eb11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:05:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/10228fc415324019a49faa54cc82d3e2 : SUCCESS in 9m 55s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/51daed7b88844976b3bf6b651f7cc42d : FAILURE in 7m 29s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/633408a1c16944e5927e2a8468113452 : FAILURE in 9m 14s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f57bb211b458452eb5a28a6ff1b3e2e6 : FAILURE in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa4020f9b9c4a0db1540ca32d73324e : SUCCESS in 10m 29s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/393d75a2ccec4981a2074a1aedb45e36 : FAILURE in 19m 03s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dbc15dbfc394e8790b64a20cc4a7856 : FAILURE in 13m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2f229a1e0114b74ad9a41e2d5dbc587 : FAILURE in 14m 11s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cb6cdd0695a4f8289ac795098a89a41 : FAILURE in 14m 32s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/ac893ce1985a436b8b9109f5e6e63658 : FAILURE in 16m 30s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6add0227ce0d48ccaae397cffe7ec2a1 : FAILURE in 12m 31s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fe9c321226741df9bc84c138e621d61 : FAILURE in 12m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d3098419ef0f7c1413481a9ad5d23107c1afa04c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-21 20:48:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5e0d07b55e83d89ba91df18a493aca6eff6d3d40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-21 21:08:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/abf1e39616c142afb8775d780a35e548 : SUCCESS in 8m 40s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/d9ff88b823084d56be601cf2c6662ab4 : FAILURE in 3m 21s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b5a0d3608f734972a8ea589d76620d0f : FAILURE in 8m 20s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a48ccc220aa4454bb0be5b1e01328a0 : FAILURE in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/915653f9c6484768804ffcdcc00d1e51 : SUCCESS in 6m 56s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/bea4154ce1d3497d95b4ee27c6f19dde : SUCCESS in 17m 49s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3d2943e96ae492d9d835290dd10ba15 : SUCCESS in 16m 29s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8bd7e75485446748660643dd2ab08e9 : SUCCESS in 16m 40s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/744663145556480a9574af3410ad238b : SUCCESS in 15m 45s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/4f162212ef5d408fbb7dc4f5e59c1129 : SUCCESS in 19m 36s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8472d1708c1e4d77b3538b662d1c731e : SUCCESS in 15m 40s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0879730d0199433ab7439dcf8c43ab6b : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2a1374fe7a23a2d4f29e308f8ec96b29bd3d0b07","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 17:11:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3b9561f8bacb147f1dbe78921ae2ce59d1af4c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 18:05:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/79c556740089421ba7a9e31cab3dd9b2 : SUCCESS in 9m 05s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/387a982668124eb5a12dd65040eb1983 : FAILURE in 6m 42s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1aac7580177c4448ac5b9c3fe4370082 : FAILURE in 5m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb2c4218943b45d4ab0a6cc476c7b1c3 : FAILURE in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85572cfe66514893b7b5bb00dad82b32 : SUCCESS in 8m 21s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/77c8646aa4db4a55a8d081400039f6db : SUCCESS in 18m 05s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a4c1eb3a8ec49bdbc9598536313577a : SUCCESS in 12m 59s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32b7a732e46f4895ba59b441f0eb96ea : SUCCESS in 14m 03s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cfef8edc78742e5a10b07b7e3104dcb : SUCCESS in 14m 33s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/ff3d667919dc4e67bd81a6e4c638ebcd : SUCCESS in 33m 37s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/401a20b9f604477b8ab932ea9fc6ee7e : SUCCESS in 14m 03s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee632fb7474a479e9572c45c61e65074 : SUCCESS in 15m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6f30edb5d2c80d4ac46470cddc1eb0814712ce13","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-18 18:09:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cafc11480802b75f4c5f726897d26bd1bbe9c9e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:36:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c2fb2e46aa264962a2d86af07fdf8c8b : SUCCESS in 5m 44s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/c1c49c99a7aa4c188e8077c884da2d20 : SUCCESS in 11m 07s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f876ed2fb44e4b4b9eeb64ca140aac1a : SUCCESS in 11m 56s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dd8541262b24d9ba7de3c8b22e9f14f : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a45eecd7a64267a2e28ffa59eb0840 : SUCCESS in 5m 31s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/547a8733f44a4424b6f78498bf7e71d3 : SUCCESS in 14m 19s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ab4481f6b494cdba07850986f524b58 : SUCCESS in 13m 58s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee3b881c4b494bf0afbecae899f8a16f : SUCCESS in 13m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83f651c7dd2849058fd0ae697a0d8dbd : SUCCESS in 13m 12s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/48e9e2251d7d40efa265d2abf678dde5 : SUCCESS in 16m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7ce86a212c04875aecd968b407d067e : SUCCESS in 15m 44s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f995bd8e8f864f97a0471c8622e2af90 : SUCCESS in 14m 44s","accounts_in_message":[],"_revision_number":2},{"id":"3c13d21aa657ceea1faad4eeab3ee6580479fd49","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-18 19:41:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"08d465bccc9dd31a8a8cf636daf224dac902963e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-05 16:56:54.000000000","message":"Patch Set 2:\n\nthis still around?","accounts_in_message":[],"_revision_number":2},{"id":"51ea77dbfdd060b78f2a9959d5eaa41b500a2137","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-10-08 10:00:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7a7b876c5c9322d30cc4bb0a6325064d3ee19143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 10:01:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9da2149f0a8bd22d18beeba13eed6c1078ff0119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 10:27:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d43c8ccdb28a4228a3954aaee966199e : SUCCESS in 13m 12s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/da822cd7d0b04b1b9f62af61695c9cde : FAILURE in 11m 20s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acde84c5213c4dec86346f6c8bbc562d : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab16ba8769b04573bad988587a7951e3 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7847882ac514ddedc380c504fe22eafd0178ce9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 10:27:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bd014050ad2f7abee77b75cbed2e66f446a8b8f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 10:29:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f7bc7466db40438369854fb5567379 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15096c16a61bb05f2d5db9e5913fc5dcd8f0f2aa","revisions":{"b79f421c20d1bd88bbf108ba074e938bac42511c":{"kind":"REWORK","_number":1,"created":"2020-05-28 23:30:00.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/25/731625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/25/731625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/731625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/731625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/731625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/731625/1"}}},"commit":{"parents":[{"commit":"69103ff82ba7acaa6b51ffce96d1f67e2b398664","subject":"Merge \"Update centos7 lxc image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/69103ff82ba7acaa6b51ffce96d1f67e2b398664"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-04-10 04:47:58.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:29:51.000000000","tz":-300},"subject":"ensure 20listchanges is not in debian container","message":"ensure 20listchanges is not in debian container\n\n20listchanges has apt call apt-listchanges\napt-listchanges is not in the base container so we get a failure\n\nChange-Id: I4b74670edd7b4ae9710dbb39deb82c0775a1ff66\nCloses-Bug: #1839535\n(cherry picked from commit 40535a00d389fd5b67822c63a98eedd95b6cab93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b79f421c20d1bd88bbf108ba074e938bac42511c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b79f421c20d1bd88bbf108ba074e938bac42511c"}]},"branch":"refs/heads/stable/train"},"15096c16a61bb05f2d5db9e5913fc5dcd8f0f2aa":{"kind":"REWORK","_number":2,"created":"2020-05-28 23:43:53.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/25/731625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/25/731625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/731625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/731625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/731625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/731625/2"}}},"commit":{"parents":[{"commit":"69103ff82ba7acaa6b51ffce96d1f67e2b398664","subject":"Merge \"Update centos7 lxc image\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/69103ff82ba7acaa6b51ffce96d1f67e2b398664"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-04-10 04:47:58.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:43:45.000000000","tz":-300},"subject":"ensure 20listchanges is not in debian container","message":"ensure 20listchanges is not in debian container\n\n20listchanges has apt call apt-listchanges\napt-listchanges is not in the base container so we get a failure\n\nChange-Id: I4b74670edd7b4ae9710dbb39deb82c0775a1ff66\nCloses-Bug: #1839535\n(cherry picked from commit 40535a00d389fd5b67822c63a98eedd95b6cab93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/15096c16a61bb05f2d5db9e5913fc5dcd8f0f2aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/15096c16a61bb05f2d5db9e5913fc5dcd8f0f2aa"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
