)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~731628","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I7c49f0cd7abb85e725296b17ecb88f34b50affed","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"master-disable-listchanges","hashtags":[],"change_id":"I7c49f0cd7abb85e725296b17ecb88f34b50affed","subject":"move listchanges removal to where it\u0027s needed","status":"MERGED","created":"2020-05-28 23:41:46.000000000","updated":"2020-08-04 18:18:47.000000000","submitted":"2020-08-04 18:17:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731628-1596565023332-bf775c37","meta_rev_id":"e69e1a50abd77b86109abec783228086c9cb17d7","_number":731628,"virtual_id_number":731628,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 18:17:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-08-04 16:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-04 17:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-04 17:59:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 16:28:20.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-04 17:59:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-04 18:17:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db8a9f4008e198f2ecf2a4bad0b0b98bc9b286cd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 23:41:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a465f7a82f9ca6f16a9b25537045a49ac8a0926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 00:04:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/70ea6d4378b7409bb2facc93e010322b : SUCCESS in 6m 04s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9a1abe23e7024bf2b97b20a5fe521cc9 : SUCCESS in 10m 24s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c515bce333fe4041a225fe4573e6b8e6 : SUCCESS in 10m 43s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/6b1bfd302e66422290226b95a6c4f592 : FAILURE in 7m 53s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77f5501fff5740b8918bd022bad31f6c : FAILURE in 8m 53s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b32d4523e46d4c0ea0475f3cf25368c1 : FAILURE in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c7c79b8a9f494f94998676554efe52 : SUCCESS in 5m 10s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/12bc3d19e1c540798188a9754a76d4c6 : SUCCESS in 16m 44s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8cc04ca469441898d8db5f8ca7b243b : FAILURE in 10m 16s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/578080578daf4ede88ecf39869835261 : FAILURE in 11m 04s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e145d360a9fc41d3b727eef167a4078f : FAILURE in 9m 09s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/a51a57a36b0947c5b273f0799308fada : SUCCESS in 17m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9209506b565f43d6acb7fa966ec2435e : FAILURE in 9m 16s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c12032a4e124676b160b29020001718 : FAILURE in 9m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d9c37de3c0b4e0d990bfa772abd975f70ed417ea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-21 20:47:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d652916c02a7d5400b65b575e186a980ce68c3ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-21 20:47:57.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"72bcab0ba9708c1c60821ce5388721ec12f57531","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-22 18:13:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6135e6a9f27a4c2083fc17941ad1e8e91d63cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 18:51:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b85bdd39ef0248f79ef61948217029d2 : SUCCESS in 10m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fb956d0dafe34d02b630a43f6beb9e49 : SUCCESS in 9m 25s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/52c19314572f432585eb791f2d5db8ac : SUCCESS in 14m 20s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8d2ac60c5f3d4a49aedf0fb0008323e8 : FAILURE in 8m 29s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68f1f5524f8b4c908a85011886ea0a3a : SUCCESS in 13m 04s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb3c023d9eab4a4c81dc241aa864c236 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e82de4a46548509b625782f504fe2f : SUCCESS in 6m 53s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/d6f02581020844a2b3364960eb2f67cd : FAILURE in 10m 11s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a8c44d5d908463f8ece53a6ef9146ae : SUCCESS in 15m 12s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eb5f9422fba4edd8e9b66b6a1254428 : SUCCESS in 16m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/913803d60e544578805afd2d5c9eb409 : SUCCESS in 15m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84d2d1a8a3c3407a91588f4839f53937 : SUCCESS in 19m 30s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa4538a5d2ae4aa4bfd3007ac89e0f06 : SUCCESS in 16m 17s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc8dc62685ef448fbaf8781aac838aeb : SUCCESS in 22m 01s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/698e5cb6c5684605be58a77f618ca9df : SUCCESS in 36m 07s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f98afe586e0945b696caeaf89b0b716e : SUCCESS in 17m 00s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a91782932194de5a83311687a01eeaa : SUCCESS in 18m 32s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0ea33ef5d1248c8ae15bd06b6b93cf4 : SUCCESS in 17m 17s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d090197ebb147c2b3043ea847b8e949 : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a4e6b049e1c89794dd7e1fae20421b7a1f9676af","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-04 16:28:20.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"28590e4fb41d844dc277ff035d70fa10545d39dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 17:02:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c12c6e9da97c416ba65177986bf2a16d : SUCCESS in 9m 45s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/04fb6e72506d40c487648b4ffcdb9e68 : SUCCESS in 12m 57s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/28c610a76fc0497aa126fcd0890138be : SUCCESS in 15m 50s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a51eb67fb2b34476b47708e743d9350b : SUCCESS in 12m 07s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d14ea0b1f374fc0b07c95cbcf67b8d5 : SUCCESS in 9m 51s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/458b61eb34d546d89265fb37851af885 : FAILURE in 12m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44481c7f0ca84515a004134a4861f7d0 : SUCCESS in 6m 52s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/b023264faee142f2a6eb1ecc61f44c2e : SUCCESS in 15m 38s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ec4adf60d2a4efe88e2dffa467545dc : SUCCESS in 14m 46s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d38262f3f9be423188baeb0ae9360a3b : SUCCESS in 17m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/734924f3fb9e4ac9b784b4126301862f : SUCCESS in 14m 34s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2ea384c6a5242bd9133b721415b3999 : SUCCESS in 18m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d2b17bce4c14425b718c308042df66b : SUCCESS in 15m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7541e1d819ba403988759adcd2d50410 : SUCCESS in 15m 22s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32ea405c5502482596432ac882ece772 : SUCCESS in 16m 46s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c337bc9fec6a429b8f7d50f060f3b304 : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0e3acf129c64d0e794784e9e87c0d1831dc91993","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-04 17:59:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63e61c1a9fcfaba0e0e06d2b07ffde487c9b20d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:00:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"804dd07a16b1f273f78adc4dfa86f5e6975def05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:17:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/889e1f2d401d40c6a32d3ef0366acdec : SUCCESS in 7m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d43d88f4c1724417a8ac5db10eafe726 : SUCCESS in 11m 59s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e2911bdecfc6457597374255e704e106 : SUCCESS in 11m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bc8318236fe49839b0da117635ef7ef : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d509dd917d5d4a58b5397e45e5aed47d : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":2},{"id":"2b59439e9d1e631b6785b59fdcf67b9062fcab56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:17:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"065bb67e675a0e9ceb1121d1763ca77df42a34dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:18:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc05a6d0e9174166826d74333a7c4819 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2aa1a6776aa729d07c05311e30f93f7e74385b97","revisions":{"6b65f9434dd60e25eb7921cf8fc3b5ed5c0c30c2":{"kind":"REWORK","_number":1,"created":"2020-05-28 23:41:46.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/28/731628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/731628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731628/1"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:40:42.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:41:40.000000000","tz":-300},"subject":"move listchanges removal to where it\u0027s needed","message":"move listchanges removal to where it\u0027s needed\n\nlistchanges config needs to be be removed before apt is run\n\nChange-Id: I7c49f0cd7abb85e725296b17ecb88f34b50affed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6b65f9434dd60e25eb7921cf8fc3b5ed5c0c30c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6b65f9434dd60e25eb7921cf8fc3b5ed5c0c30c2"}]},"branch":"refs/heads/master"},"2aa1a6776aa729d07c05311e30f93f7e74385b97":{"kind":"REWORK","_number":2,"created":"2020-06-22 18:13:12.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/28/731628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/731628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/731628/2"}}},"commit":{"parents":[{"commit":"54d1f8c4794b9d61e4d127f9ccffe5ef41f8eebf","subject":"Merge \"copy the actual keyring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/54d1f8c4794b9d61e4d127f9ccffe5ef41f8eebf"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:40:42.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-06-22 18:12:59.000000000","tz":-300},"subject":"move listchanges removal to where it\u0027s needed","message":"move listchanges removal to where it\u0027s needed\n\nlistchanges config needs to be be removed before apt is run\n\nChange-Id: I7c49f0cd7abb85e725296b17ecb88f34b50affed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2aa1a6776aa729d07c05311e30f93f7e74385b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2aa1a6776aa729d07c05311e30f93f7e74385b97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
