)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~731650","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ibf7e4cad51d0e13dbd59564ce28f0a1af0933103","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"install-libs-for-debian","hashtags":[],"change_id":"Ibf7e4cad51d0e13dbd59564ce28f0a1af0933103","subject":"install libs for needed for keystone on debian","status":"ABANDONED","created":"2020-05-29 05:23:07.000000000","updated":"2020-05-29 15:13:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7fca2c511696f5335fb45ba310e496b64de78fe1","_number":731650,"virtual_id_number":731650,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-29 05:52:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-05-29 06:14:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 05:52:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-29 06:14:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"e9efbade1f8f57d4217e7dd77a3ee75860ae7f51","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-29 05:23:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2db500a7ae8b49ed6615c7bea43b4989e36316d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 05:52:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/db700db6356c4e439e9355ee4e5bac19 : SUCCESS in 8m 49s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d99054bc0a1a4e01ade81cfe4d1e33d0 : SUCCESS in 11m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/89b89098a8704ac5a3de6acca05c5779 : SUCCESS in 18m 59s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d5a4099282144aecb797bcc67ea15653 : FAILURE in 10m 00s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbea23ccd28941f2819a43ccf288779f : FAILURE in 11m 51s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e14e1fbc99f74b7f9ac1de88f0517d48 : FAILURE in 12m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48447c62c49e45bd951db157e4bc29c6 : SUCCESS in 7m 40s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/14da0d5d9add4c0695eea818c4ef6624 : SUCCESS in 26m 41s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/697c800610f542ae9e916d80f9833913 : FAILURE in 14m 00s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba98c07a26b9413dbc23acece0096fe3 : FAILURE in 13m 40s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/197ac653ed9947a4b88ff36b6b385bfd : FAILURE in 12m 04s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/f78d0501f8ef429c81fabda0a3e19eb7 : SUCCESS in 25m 35s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/015ce420ea5640fe961ae5d5c42e2c6b : FAILURE in 14m 03s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9e09c4e0d5c45b3a16454480f430688 : FAILURE in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"75cb85a0765f56ae149ba671d5a74cfcf03d030a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-29 06:14:19.000000000","message":"Patch Set 1: Code-Review-1\n\nI believe that would be these https://github.com/openstack/openstack-ansible-os_keystone/blob/master/vars/debian.yml#L25","accounts_in_message":[],"_revision_number":1},{"id":"0e189619bd42a789cbf558058d06f180145a20c4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-29 15:13:36.000000000","message":"Abandoned\n\nissue was the repo_server, review there fixes this","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1a6ec7743659dfdaf09a6a2e9e8e1fa9ad975a2","revisions":{"e1a6ec7743659dfdaf09a6a2e9e8e1fa9ad975a2":{"kind":"REWORK","_number":1,"created":"2020-05-29 05:23:07.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/50/731650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/50/731650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/731650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/731650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/731650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/731650/1"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-29 05:22:37.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-29 05:22:37.000000000","tz":-300},"subject":"install libs for needed for keystone on debian","message":"install libs for needed for keystone on debian\n\nChange-Id: Ibf7e4cad51d0e13dbd59564ce28f0a1af0933103\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e1a6ec7743659dfdaf09a6a2e9e8e1fa9ad975a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e1a6ec7743659dfdaf09a6a2e9e8e1fa9ad975a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
