)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~735612","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I8e1fdc42a604784f83fe56467f736424c1abe9d8","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"I8e1fdc42a604784f83fe56467f736424c1abe9d8","subject":"WIP - Remove machinectl from LXC cache prep","status":"ABANDONED","created":"2020-06-15 15:13:33.000000000","updated":"2020-06-16 05:37:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0926f91b6df0af9296db51d3e1969dab7e840ea9","_number":735612,"virtual_id_number":735612,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-15 15:53:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 15:53:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1eb2d8d7413c8fb46a79c536d86e493e04460e1b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-15 15:13:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38a1cfeac9e5fdeb05aa9327161037a57572a2a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 15:53:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d41020a3720140feb0224b4e7c15ea9d : FAILURE in 11m 17s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7ddb9eac036b4f31abdca4cc97ca8e9d : SUCCESS in 13m 57s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1caa25b7a86b4201b481f56e6f197a7f : SUCCESS in 17m 20s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/535654681cea4c3c9c5c7ba31e84f80a : FAILURE in 17m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e596b61c084e4a999663e13c0751989e : SUCCESS in 12m 45s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a8aa0e0d78447f5b407f85eb3937a23 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee8401e1f434e84818bf41a6f81e5d0 : SUCCESS in 10m 06s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/f8d87d2cd6a04b8fbbff18131057d7fe : SUCCESS in 21m 08s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a72a83f72a794a6ea8a96d10573e99d4 : SUCCESS in 16m 57s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/792fe2194a1549378c10ff340af31254 : SUCCESS in 16m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81cf8e41f2864d48ba876383d1f5a7de : SUCCESS in 21m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d34016fd4fba423582a5445de580493b : SUCCESS in 22m 33s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a80959054cb4ac0878ba1cbefc650d4 : FAILURE in 15m 11s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c78a7a909c0f402dac1ced2b78bdca05 : FAILURE in 18m 19s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/dbbe1ff0b32840dc99a810bdbbf9c5c1 : FAILURE in 17m 03s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/969ce5b2f6904db19c8b8c6f3c91518d : SUCCESS in 18m 52s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40f847da4e2a43b28a99a50314b28027 : SUCCESS in 20m 03s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37c465d8127247df8b2ff12e64015242 : SUCCESS in 17m 52s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5df2822617014953bf5bb30304f12a52 : SUCCESS in 24m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ef6050dff66cf4626e9b6100ec8daa5528947706","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-16 05:37:13.000000000","message":"Abandoned\n\nDuplicate","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6df51f9c9c07be76ccfdce01f106d7ad1a49bc7d","revisions":{"6df51f9c9c07be76ccfdce01f106d7ad1a49bc7d":{"kind":"REWORK","_number":1,"created":"2020-06-15 15:13:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/12/735612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/12/735612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/735612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/735612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/735612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/735612/1"}}},"commit":{"parents":[{"commit":"e484938285960d2e4b727f9b0561b18bf5c3b748","subject":"[WIP] centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e484938285960d2e4b727f9b0561b18bf5c3b748"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-15 15:12:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-15 15:12:56.000000000","tz":60},"subject":"WIP - Remove machinectl from LXC cache prep","message":"WIP - Remove machinectl from LXC cache prep\n\nmachinectl is only used to store the image during\ninitial cache preparation and is unrelated to the\nbacking store used by LXC.\n\nThis patch removes the use of machinectl and btrfs\nwhich makes the lxc_hosts role portable to centos-8\nwhich does not have btrfs.\n\nChange-Id: I8e1fdc42a604784f83fe56467f736424c1abe9d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6df51f9c9c07be76ccfdce01f106d7ad1a49bc7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6df51f9c9c07be76ccfdce01f106d7ad1a49bc7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
