)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~735780","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ifdeb447e1ef000dbe83394f6e5b0ed3c7afc84c5","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"Ifdeb447e1ef000dbe83394f6e5b0ed3c7afc84c5","subject":"Download yum keys to host before installing","status":"MERGED","created":"2020-06-16 06:48:17.000000000","updated":"2020-06-24 13:23:50.000000000","submitted":"2020-06-16 12:39:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735780-1592311172360-a663f06a","meta_rev_id":"a990b68a4a0af634f1098147de03834eb35865bc","_number":735780,"virtual_id_number":735780,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 12:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-16 09:56:40.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-06-16 12:17:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-06-16 11:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-06-16 12:17:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 11:52:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-16 12:17:48.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-16 12:39:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbdf97ecd5ecf1177ab7ed173a43e6414c6433bc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-16 06:48:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7777d4d6f591d8e32769f7367d7cf9814fbead21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 07:32:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/492c2db597234c028361378cc53e1886 : SUCCESS in 11m 52s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/88d6a98bb9fa4e04ab89f14e0595199d : SUCCESS in 15m 02s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3b2a34ed1fd34c829a09a1de2fe0f243 : SUCCESS in 16m 38s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7961ebd91698449baf89050084a97e92 : FAILURE in 11m 31s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbc72e2f442a4af19fb870b3a7583e6d : SUCCESS in 14m 54s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe1d3110e1d84f27b12f3ad10bfc17b2 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01947a411b804e9b82b8e94f3741f1f6 : FAILURE in 8m 26s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/1ca24708095b48f1b94fdfc38056a733 : SUCCESS in 23m 05s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d44d969be014860b8ffbe86ff4b2759 : SUCCESS in 18m 59s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87280f05ab324da7b45a664504065040 : SUCCESS in 15m 41s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b88eda07e39a49bc92db4d15510740ec : SUCCESS in 17m 20s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e628bc4c886b47dd8785c7c005e770e6 : SUCCESS in 18m 16s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9a0ad9a6e9144c99cffaa9f11d67cfa : SUCCESS in 16m 05s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dac92cb973242beb48536b2aecc9ca8 : SUCCESS in 16m 49s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/47d6a9d1ce57400a8b4080393e304541 : SUCCESS in 19m 18s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7020acb8e2942d98de7beab69e14d17 : SUCCESS in 19m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/002c54f36291446f89587e0795fb748a : SUCCESS in 21m 42s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd015d04f2b4499496853eb5c8e37e1d : SUCCESS in 18m 54s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc38fe0c29f64423810a465f2bb578b9 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ff234a352acf267bd018f6f84ba080bba6833382","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-16 08:11:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e73f95650f679d13283726470090682f73b6cd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 08:52:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e3fe6f2445d44f63a34b06be5f45a62c : SUCCESS in 9m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/656bd78beb4d41c890611c06e10b6024 : SUCCESS in 11m 41s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c8579c70165f46978a02044518cd2cda : SUCCESS in 13m 45s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/aaca5f6aa2e248738f253f79f97b10de : FAILURE in 11m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5c041bd8d3e4423832d649f59a9cf7d : SUCCESS in 11m 31s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45e74519ae2c49d8b26783c25fdf39b4 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402c10effb2d41fc804db58e3f94027d : FAILURE in 9m 23s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/bd939b2e32c54b4694f21ee6f37397df : SUCCESS in 27m 03s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c5e0ce0375f4584a26a7a77e25c47d3 : SUCCESS in 23m 28s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75208fc1e89f4c348ce46441065700f5 : SUCCESS in 17m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fa3174777d84304b3732fc5fb3e015a : SUCCESS in 16m 57s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7087e2f6ba74c7894e7aa40d4d0af32 : SUCCESS in 17m 21s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4ab20757acd438193b99ae944d5482f : SUCCESS in 16m 55s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb29f4bb45204ea3a04a52960cbfa779 : SUCCESS in 18m 28s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/257e95b601184c89af78c849070e7b42 : SUCCESS in 20m 26s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76850ee887b843b8ae3a6cd42a7ce7df : SUCCESS in 25m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/272e1d1219d841749c36aa337a3655ec : SUCCESS in 20m 32s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/226ba1d9dec448bf9e7933d3c8826fa9 : SUCCESS in 23m 26s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cbcf9ecacc04a25a8b6ff64d3078803 : SUCCESS in 21m 31s","accounts_in_message":[],"_revision_number":2},{"id":"943eef02416b8ae045194aaf140d5127d5f6d2c9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-16 09:56:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"59c468f38ff35cb1e7c9632edbc5c8b386359c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 10:32:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d48ebba2dab94f3887a77e40ec948d5f : SUCCESS in 11m 04s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/781afb208c954ca5828578ffa04692a2 : SUCCESS in 11m 15s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0e55648ad4de47c79ca0f59559f119a1 : SUCCESS in 13m 28s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/5239b9a83d474de397ea95c3509fedfa : FAILURE in 11m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc135a1ce2c64816955b693ab92c9ad3 : SUCCESS in 14m 13s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c1d3cc1bfcd4ba5a7624353f7ef6b0f : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdc47eadb494816b80d150f7b8b64f3 : SUCCESS in 8m 56s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/9bbe5edc703343c4a553c18beee36b54 : SUCCESS in 19m 15s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb025d5b96ed4406a3022303fe935b93 : SUCCESS in 14m 08s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a20eee6064748e583de59f4ba102f3b : SUCCESS in 17m 20s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1ad268a9dd24065b0b43b6b19c76586 : SUCCESS in 19m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1da68895c95c40c6a664d7f221eaf270 : SUCCESS in 17m 25s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/310652a79d884240ad4124c66b7029b3 : SUCCESS in 16m 39s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49fec301fe034ea8b23dd29037347c0f : SUCCESS in 17m 15s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/5bffb14b9bbf4b22a9d6fb03d3dd6bc7 : SUCCESS in 18m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f78239b977bb49daaa712bb497cf9ecf : SUCCESS in 18m 03s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d968c4be0d814458825a2161221be187 : SUCCESS in 19m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eca3e0a8e4a94e6bb78948ccae1fac9d : SUCCESS in 16m 22s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/621657aef9bd479194259634bf39ba39 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":2},{"id":"3232a5b7368396fa9a79b1efa7385ab2fc1bf907","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-16 11:52:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0ec7f69e9a22557c14f9fb5a410205c3161b3c26","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-16 12:17:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b60892bf85d1faa9316123aa4b20c0f2a30872a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:17:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60c07f6741b6d392b1bdd1daff81124dce348f01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:39:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ad1c52f65d1a4250a4f2ed12eabf9cb4 : SUCCESS in 8m 56s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2409ee8ed7dd49ddb3b2c8d26b579653 : SUCCESS in 12m 04s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b593fd9ae01345b8bbff85ac5ba8f461 : SUCCESS in 19m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/387b940909ed4246842f140c644cb1e2 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92eb2cba9d44306ad1b54a82efd1c52 : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ffe6373032386b5f35f9c60703b581fe3d560a75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:39:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"87f6db77a38195c497acade80e256d5e2bef9bf2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:43:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a0db37126ed4f5fa41745dfec15177c : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a744c830d103f76745aad692bfeadd72832b6356","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 13:23:50.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 2e57371ec4cd7b41d2cb00dd979534ac287dde4f","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"295c3c0a935b72d15211882fc5253ba44c26d917","revisions":{"37c751cad41ba785773373029f674980cf92465f":{"kind":"REWORK","_number":1,"created":"2020-06-16 06:48:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/80/735780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/80/735780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/735780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/735780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/735780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/735780/1"}}},"commit":{"parents":[{"commit":"37f3a9fe9e9154468934435d78ac77423031077c","subject":"Merge \"Add functional tests for ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/37f3a9fe9e9154468934435d78ac77423031077c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-16 05:56:46.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-16 06:47:11.000000000","tz":60},"subject":"Download yum keys to host before installing","message":"Download yum keys to host before installing\n\nThe LXC image prep script copies the contents of /etc/pki/rpm-gpg to\nthe container image so that these keys can be used inside the container.\n\nImportantly, /etc/pki/rpm-gpg is only a staging area where keys are\nkept on the filesystem and is not the actual set of keys imported\ninto the package manager database.\n\nFor the EPEL key to be properly copied into the LXC container image\nit must first be staged into the host /etc/pki/rpm-gpg directory\nand then installed using the rpm_key module. If the key is installed\ndirectly using rpm_key then it is not available as a file to copy\ninto the container image.\n\nDepends-On: https://review.opendev.org/735289\nChange-Id: Ifdeb447e1ef000dbe83394f6e5b0ed3c7afc84c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/37c751cad41ba785773373029f674980cf92465f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/37c751cad41ba785773373029f674980cf92465f"}]},"branch":"refs/heads/master"},"295c3c0a935b72d15211882fc5253ba44c26d917":{"kind":"REWORK","_number":2,"created":"2020-06-16 08:11:06.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/80/735780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/80/735780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/735780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/735780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/735780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/735780/2"}}},"commit":{"parents":[{"commit":"37f3a9fe9e9154468934435d78ac77423031077c","subject":"Merge \"Add functional tests for ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/37f3a9fe9e9154468934435d78ac77423031077c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-16 08:06:08.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-16 08:06:08.000000000","tz":60},"subject":"Download yum keys to host before installing","message":"Download yum keys to host before installing\n\nThe LXC image prep script copies the contents of /etc/pki/rpm-gpg to\nthe container image so that these keys can be used inside the container.\n\nImportantly, /etc/pki/rpm-gpg is only a staging area where keys are\nkept on the filesystem and is not the actual set of keys imported\ninto the package manager database.\n\nFor the EPEL key to be properly copied into the LXC container image\nit must first be staged into the host /etc/pki/rpm-gpg directory\nand then installed using the rpm_key module. If the key is installed\ndirectly using rpm_key then it is not available as a file to copy\ninto the container image.\n\nDepends-On: https://review.opendev.org/735289\nChange-Id: Ifdeb447e1ef000dbe83394f6e5b0ed3c7afc84c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/295c3c0a935b72d15211882fc5253ba44c26d917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/295c3c0a935b72d15211882fc5253ba44c26d917"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
