)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~737346","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~I7c49f0cd7abb85e725296b17ecb88f34b50affed","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","topic":"remove-listchanges-ussuri","hashtags":[],"change_id":"I7c49f0cd7abb85e725296b17ecb88f34b50affed","subject":"move listchanges removal to where it\u0027s needed","status":"MERGED","created":"2020-06-22 18:16:14.000000000","updated":"2020-08-05 10:44:13.000000000","submitted":"2020-08-05 10:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737346-1596624084372-60d190a1","meta_rev_id":"6cb222fdc9883608c2451f53f7d3c21e2bbd031c","_number":737346,"virtual_id_number":737346,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-05 10:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-08-05 10:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-05 07:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-08-05 10:15:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 07:06:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-05 10:15:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-05 10:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f3d806bd18d20a8a85efd488917c34bc4043055","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-22 18:16:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"148dbc63a99a99010467a49b25daa05d6200607e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-22 18:20:24.000000000","message":"Abandoned\n\ndid not have cherry-pick message","accounts_in_message":[],"_revision_number":1},{"id":"d34d230bb89ba33ed580cafa50bba25e7ce36539","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-22 18:25:17.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a61608e1b51d34f416528123312d87425b37b7bd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-22 18:25:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b090f0569344c8b20270ed88fc022fc6eb43af37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 18:51:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a094c02e90ee45b3a31f22492bf06b68 : SUCCESS in 6m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/21f66a2981f94749bae0afa9b7848b8b : SUCCESS in 20m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0a92192850ff422abaada510971f14d5 : SUCCESS in 14m 18s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/0ab0a3e3b5e3436cb5b339d42d1bbf94 : FAILURE in 15m 02s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee6f3cf63adb41e7b43e2ccd5413d037 : SUCCESS in 12m 26s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90cb7297c2464e1aaf06a7ae531d3bb7 : FAILURE in 10m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d716f4f040b4f9fa118486d9d7f57be : SUCCESS in 7m 22s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/7a3fdc911003460683bf189aa1932b52 : SUCCESS in 20m 22s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b543ed42e96140219669e1bff3fc97b3 : SUCCESS in 17m 23s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02a3c934ba664202b59b4abdf72d800a : SUCCESS in 12m 25s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcad42ee3a164e6cbac9f1e5dffb6da1 : SUCCESS in 16m 59s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/c5a9f5412060441ba17078620321491e : SUCCESS in 20m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c717076d4f44683836327b754f51746 : SUCCESS in 17m 46s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8428167f35840d89d17b3d3ab4892eb : SUCCESS in 22m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3dfe9365d920a776712f578cbcfbfffcde2c1833","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 10:52:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"720f551de3cba225d03998999ba3742d65af4486","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-04 16:27:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f7a8948f0593fe7f9fd8afc9b1d0020bed18fce2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-04 16:29:16.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7af74564b4e1edc6f5101c48862350b0745b25ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 17:04:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2228432eaf4245b08c2134a1750fd028 : SUCCESS in 7m 48s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/86fc1faa87a34f7c853b2a354b411d3e : SUCCESS in 11m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5fd22c6036ab452f8bd1f72723fdb5c3 : SUCCESS in 12m 38s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/35002a2a342048f1b5bed6099e36fa8e : SUCCESS in 13m 37s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b003aecf3a94aec8506da8a49986568 : SUCCESS in 9m 52s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed14ff49cda74edcb06a3262edf679e7 : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f4bb0b8b9440d79697c8ede8c9239a : SUCCESS in 6m 30s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/a0c044f398634a97873d65a721662024 : SUCCESS in 19m 08s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d6cb5e2d5394fd9bc0f54208ef0081c : SUCCESS in 16m 52s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83c95490ff71437891f66e818ada1462 : SUCCESS in 14m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2553f7452a0246cb97b067df1c905298 : SUCCESS in 16m 34s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adb933e9d286433a88c33ca6505b2d3c : SUCCESS in 16m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71a47a9e34124e028ecf141955b09ad4 : SUCCESS in 19m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/517181e5a47045cfa6fa6aa76743d84c : SUCCESS in 17m 11s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d06c59d7370044ce819bc82232ce137c : SUCCESS in 18m 13s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/391be1cd944b477f95af188dadd563a2 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":4},{"id":"b120e20953cdeef2feaf9af4abf43d1d04be1d84","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-05 07:06:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"94f1d29122c3418dfae611e9c31b238532f5a622","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-05 10:15:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"01fec6593efd1ef52452365ffeb05aababbc3331","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:19:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"317b0eff4ede55c005609f3438926d2ccdafc8ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:41:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e7f74339fde84f8db1b3219bd4ec973d : SUCCESS in 6m 27s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c9521cb804ca410ea09ff6b52a254e9e : SUCCESS in 9m 19s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/20637fdc51464887a488e737ae62c83b : SUCCESS in 10m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ddc596100704f86a99f943ad0994003 : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56208146d06346b6bb048e4fa12956e2 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":4},{"id":"376dba5501372109f94641ae07478dd80f23326c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:41:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"486f9ef6f669c2ecab36243a84d07cf81dd1354b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:44:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088dd544c09e409e829482d84c5b0e80 : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cb4f10bdf1692be78790e06b21a3f60f7d337f8c","revisions":{"78c0144534a990edc0169a0b5f3150f9d3973685":{"kind":"REWORK","_number":1,"created":"2020-06-22 18:16:14.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/46/737346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/46/737346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/1"}}},"commit":{"parents":[{"commit":"2171a3b529236455a4ead95ec55b183f2b335eec","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2171a3b529236455a4ead95ec55b183f2b335eec"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:40:42.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-06-22 18:16:04.000000000","tz":-300},"subject":"move listchanges removal to where it\u0027s needed","message":"move listchanges removal to where it\u0027s needed\n\nlistchanges config needs to be be removed before apt is run\n\nChange-Id: I7c49f0cd7abb85e725296b17ecb88f34b50affed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/78c0144534a990edc0169a0b5f3150f9d3973685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/78c0144534a990edc0169a0b5f3150f9d3973685"}]},"branch":"refs/heads/stable/ussuri"},"3b1d59c7f5c384cd49c171521a73a7813f34dfb8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-22 18:25:22.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/46/737346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/46/737346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/2"}}},"commit":{"parents":[{"commit":"2171a3b529236455a4ead95ec55b183f2b335eec","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2171a3b529236455a4ead95ec55b183f2b335eec"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:40:42.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-06-22 18:22:05.000000000","tz":-300},"subject":"move listchanges removal to where it\u0027s needed","message":"move listchanges removal to where it\u0027s needed\n\nlistchanges config needs to be be removed before apt is run\n\nChange-Id: I7c49f0cd7abb85e725296b17ecb88f34b50affed\n(cherry picked from commit 2aa1a6776aa729d07c05311e30f93f7e74385b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3b1d59c7f5c384cd49c171521a73a7813f34dfb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3b1d59c7f5c384cd49c171521a73a7813f34dfb8"}]},"branch":"refs/heads/stable/ussuri"},"3c29c0a6c2a76686fb36d2a9fe92c6d7dcd318f9":{"kind":"REWORK","_number":3,"created":"2020-08-04 16:27:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/737346/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/46/737346/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/3"}}},"commit":{"parents":[{"commit":"102b7780e85a354dd8ceff481d31519b1ff7bbba","subject":"Add remote_src parameter to ansible unarchive module for lxc_hosts task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/102b7780e85a354dd8ceff481d31519b1ff7bbba"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:40:42.000000000","tz":-300},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-04 16:27:06.000000000","tz":180},"subject":"move listchanges removal to where it\u0027s needed","message":"move listchanges removal to where it\u0027s needed\n\nlistchanges config needs to be be removed before apt is run\n\nChange-Id: I7c49f0cd7abb85e725296b17ecb88f34b50affed\n(cherry picked from commit 2aa1a6776aa729d07c05311e30f93f7e74385b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3c29c0a6c2a76686fb36d2a9fe92c6d7dcd318f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3c29c0a6c2a76686fb36d2a9fe92c6d7dcd318f9"}]},"branch":"refs/heads/stable/ussuri"},"cb4f10bdf1692be78790e06b21a3f60f7d337f8c":{"kind":"REWORK","_number":4,"created":"2020-08-04 16:29:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/737346/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/46/737346/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/737346/4"}}},"commit":{"parents":[{"commit":"102b7780e85a354dd8ceff481d31519b1ff7bbba","subject":"Add remote_src parameter to ansible unarchive module for lxc_hosts task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/102b7780e85a354dd8ceff481d31519b1ff7bbba"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-28 23:40:42.000000000","tz":-300},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-04 16:29:13.000000000","tz":0},"subject":"move listchanges removal to where it\u0027s needed","message":"move listchanges removal to where it\u0027s needed\n\nlistchanges config needs to be be removed before apt is run\n\nChange-Id: I7c49f0cd7abb85e725296b17ecb88f34b50affed\n(cherry picked from commit 2aa1a6776aa729d07c05311e30f93f7e74385b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cb4f10bdf1692be78790e06b21a3f60f7d337f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cb4f10bdf1692be78790e06b21a3f60f7d337f8c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
