)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~737769","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~Ie94afc5a3c794b1c2be266b6642bc9c74b533287","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","topic":"osa-focal-stable/ussuri","hashtags":[],"change_id":"Ie94afc5a3c794b1c2be266b6642bc9c74b533287","subject":"Stop installing python2.7 on debian/focal","status":"MERGED","created":"2020-06-24 13:05:37.000000000","updated":"2020-06-25 11:17:06.000000000","submitted":"2020-06-25 11:14:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737769-1593083682750-5ae3cf4f","meta_rev_id":"8bd46c5cd9965fff50ed18475607b9e4eadcffd6","_number":737769,"virtual_id_number":737769,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 11:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-06-25 09:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-25 09:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-06-25 09:16:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-25 10:56:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 09:16:15.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-25 11:14:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e13e2a56e5da2b57a73be8ddfb2e8c3a6d54a53","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 13:05:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"257912b624acb958a8c90a93fde3ddfdc4908257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 14:10:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/53764526fb914c8eb89f5458f4fd4e55 : SUCCESS in 6m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ac7435effc564ff7a29531dac8a9895b : SUCCESS in 8m 58s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b204e90ad5cb4419b4d03b70436c2908 : SUCCESS in 10m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/96415b19b23c44bfa5778df32c912cbd : FAILURE in 8m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbb7abedac44491d814249ed38264ea3 : SUCCESS in 11m 05s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7bb7a7c9f3e458c9ec984984c2eb68d : FAILURE in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56025e6afa304779bc7d56e0765675ec : SUCCESS in 5m 26s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/a50b198b19bb4a0dac4ec8583307060b : SUCCESS in 18m 43s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d607cc9d62cf465298df1eab2d2f7164 : SUCCESS in 14m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa3b44b29ece4f68ba55b46096343a4d : SUCCESS in 15m 13s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f16c13bb28744a9a51a72739f35cce2 : SUCCESS in 14m 08s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/745eeb087913498b87440210dc7f9c67 : SUCCESS in 17m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/090bfcd34e0d43108c0a0a3e3f05faf4 : SUCCESS in 14m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebec474d6efd4c0e839c2cfa824c1546 : SUCCESS in 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d28ecad8f6cc9e28460c830833bdcad6022ced6a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 14:19:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d84acb78385550982b79cf8f6beae819b0fca76d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 06:47:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"09b17f1549390ae9a3a62e7a19d75cbdbb2f2036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 07:13:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1dd613014a6b472e8cbbf148d3430945 : SUCCESS in 11m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7fdaf1fdc6ee4b458239f368f7639233 : SUCCESS in 14m 07s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/632fe3b18ae946d4b045fb3fd4a936b6 : SUCCESS in 13m 17s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/9f0ca5fe6c724011a5b3d7c4a3a1f86f : FAILURE in 10m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c47db6135a064343844043b3e4dbce76 : SUCCESS in 12m 23s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/100575bd537946fb9648e036a881dc4b : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59b1c8f6454d495ab6ec4afe4408f049 : SUCCESS in 11m 37s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/8cdc2afeca6b424ebb556f7860d0e696 : SUCCESS in 17m 47s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32a3eab0a4b64d6c8e27fe982bedff3d : SUCCESS in 17m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39c68b788a564a95982d1b6bd183e61b : SUCCESS in 17m 11s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bde4dc6b2cf41e188a2e5650fdfa64f : SUCCESS in 16m 23s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/9e1f9a35c1474237934a0b3e88a17375 : SUCCESS in 17m 00s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ef9fde4dce84833910ae0fe3452e7e1 : SUCCESS in 16m 10s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c806234c5d44cfd97d64957187ce1dd : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":2},{"id":"4c4110b517ba3b6690a8c8a41160c02297f9f591","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 09:14:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0bf676cb8755afb25140205be6627026e6bacff0","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-25 09:16:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7968feee60bb25d4b90246361faf951d606efe9e","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-25 09:16:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8eef111034f9ebdc4e2081f5aecc811aded46178","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 10:56:59.000000000","message":"Patch Set 2: Workflow+1\n\npush to gate","accounts_in_message":[],"_revision_number":2},{"id":"1fe703296dfb5c852e02acf69dcebf9516790180","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:57:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9bcee33c24187933c60aa394672f33fd7936ab68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:14:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fa97d65dca367aaa655bf6399973725e7181ba27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:14:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f2c80675e8b3493db98867916e85afe1 : SUCCESS in 6m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/30c3b1fd05084a82823c6aa522853466 : SUCCESS in 12m 28s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0ae2821b34ce41449c2f5cf4bd8734b0 : SUCCESS in 12m 16s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ac873dd7ec9432984c4db576ebb3fd8 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0cffecfc5704f01b28f6f2826f69ece : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"df4bf052236b5ea4af9433d75147c4c3e37aea0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:17:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b742cf22c88f47e7832248886a4aaa85 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"efbc473b01dcc7cc4b265132f93437e479ea0cdc","revisions":{"3d1034dc9837a7c0fa80c3488486f11454e24b7c":{"kind":"REWORK","_number":1,"created":"2020-06-24 13:05:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/737769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/69/737769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/737769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/737769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/737769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/737769/1"}}},"commit":{"parents":[{"commit":"2171a3b529236455a4ead95ec55b183f2b335eec","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2171a3b529236455a4ead95ec55b183f2b335eec"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-28 08:05:02.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-24 13:05:37.000000000","tz":0},"subject":"Stop installing python2.7 on debian/focal","message":"Stop installing python2.7 on debian/focal\n\nWe still require py2 and the lxc library on the host for bionic as\nthe bionic CI image still has python2 at /usr/bin/python and ansible\ninterpreter discovery picks that when targetting localhost.\n\nChange-Id: Ie94afc5a3c794b1c2be266b6642bc9c74b533287\n(cherry picked from commit dfc9502e225efbb0d395f2e450d1474f6e535483)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d1034dc9837a7c0fa80c3488486f11454e24b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d1034dc9837a7c0fa80c3488486f11454e24b7c"}]},"branch":"refs/heads/stable/ussuri"},"efbc473b01dcc7cc4b265132f93437e479ea0cdc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-25 06:47:46.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/737769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/69/737769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/737769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/737769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/737769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/69/737769/2"}}},"commit":{"parents":[{"commit":"2171a3b529236455a4ead95ec55b183f2b335eec","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2171a3b529236455a4ead95ec55b183f2b335eec"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-28 08:05:02.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-25 06:47:40.000000000","tz":0},"subject":"Stop installing python2.7 on debian/focal","message":"Stop installing python2.7 on debian/focal\n\nWe still require py2 and the lxc library on the host for bionic as\nthe bionic CI image still has python2 at /usr/bin/python and ansible\ninterpreter discovery picks that when targetting localhost.\n\nDepends-On: I79b2c72132d7a8e8fe1e06b382ef4de03c05d8b4\nChange-Id: Ie94afc5a3c794b1c2be266b6642bc9c74b533287\n(cherry picked from commit dfc9502e225efbb0d395f2e450d1474f6e535483)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/efbc473b01dcc7cc4b265132f93437e479ea0cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/efbc473b01dcc7cc4b265132f93437e479ea0cdc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
