)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~737770","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~Ifcb09830dc6d4c9f06db907a4b646a230b1ad938","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","topic":"osa-focal-stable/ussuri","hashtags":[],"change_id":"Ifcb09830dc6d4c9f06db907a4b646a230b1ad938","subject":"Add functional tests for ubuntu focal","status":"MERGED","created":"2020-06-24 13:05:57.000000000","updated":"2020-06-25 11:26:22.000000000","submitted":"2020-06-25 11:23:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737770-1593084181971-dd5307e8","meta_rev_id":"84ca5c02841c05f64eed0de03f510c8413b379e9","_number":737770,"virtual_id_number":737770,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 11:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-06-25 11:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-25 09:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-06-25 11:04:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 11:04:39.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-25 11:23:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c081f0fa612dcafde819c0e2aa771e6842c8bde2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 13:05:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3039e53260d29c0d9e10fac0b0b59498d048aea1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 13:07:54.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5ae6536b179e1d202c34b73281b7e7ce23dc91cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 13:44:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0f863e75c0e148f9867bcfccb174eafa : SUCCESS in 6m 44s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/db5fb4010e544960856f5c8fe35f5ab4 : SUCCESS in 8m 08s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2c0d20b52b564355a27925775143dc90 : SUCCESS in 10m 12s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f19eda5c683e4a9ebabdbdfead367878 : FAILURE in 8m 22s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/042e2c6b74214352b760d15596e878cf : SUCCESS in 11m 30s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c1b70261a3e4bfd84321e3dcdafe204 : FAILURE in 10m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c551c7bc1740eea5aed67589885033 : SUCCESS in 5m 39s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/533143abaf804c0d80456da6bcc8cef6 : SUCCESS in 18m 11s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50a48a48bca2471b8b8cd5145fde12e0 : SUCCESS in 15m 30s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9f0617202ba44509eef385b3fd368f2 : FAILURE in 9m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81a2559acdf9474db45d9e4aa7b99061 : SUCCESS in 17m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/327c7750fee342c894a9b4eb98ad8a9f : FAILURE in 6m 52s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed0a424f556d4304be829067ea76fd1c : SUCCESS in 14m 57s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e80c1d5eeee4b73b4c8ea183adf6b34 : FAILURE in 9m 18s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/0af088b8024e43d7ae377bbb7a3acd3e : SUCCESS in 14m 39s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dc30cc75bf84f64bf07c6546d9904b5 : SUCCESS in 13m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cd3d98150564e469e92a8a1706aee5d : FAILURE in 12m 10s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a256f17b0dde4f61a94a8c89633f0149 : SUCCESS in 12m 22s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c9455f7e7ec4da2b4b2e37f6625fea9 : FAILURE in 10m 03s","accounts_in_message":[],"_revision_number":2},{"id":"16485f954096e14ef4d065392299fbee0bf34477","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 06:49:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"36cb75acf20084af7baae63900988b368a0cf89c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 07:42:40.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9b9c1038993c9756599c6ddd361cb8bd43d2f01c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:11:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/40e91fe00a36463c9d782d05973e2862 : SUCCESS in 8m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ee283109e87a46f2b678229fc51f6be9 : SUCCESS in 9m 03s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7820274d23804b9bbb2ec2378633b3a9 : SUCCESS in 13m 18s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f145bc4bf85c46aa9e48ce438138c77c : FAILURE in 10m 00s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9115296b30f74198abe145f7ea7ecc46 : SUCCESS in 12m 20s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/022b0ea4a9d743f08a4679bd36960843 : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5345000af3514c03b6a6bdcf4be789c1 : SUCCESS in 8m 22s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/ea3963935b68484d8ca78a5d439073fd : SUCCESS in 21m 58s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/594583f24d0a4853b97e08e3b19c319b : SUCCESS in 15m 07s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90fe55d52890470e85f33e377be6fadb : SUCCESS in 17m 41s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e12cd241f6fe4a85b9db49d01b666d9e : SUCCESS in 18m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b371e901e18f442a87b574e0094bdfa1 : SUCCESS in 19m 12s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3505a66b0bee4d18b4f8f4729745cf1a : SUCCESS in 17m 24s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/793820e3ca3646758756a08a148434bf : SUCCESS in 22m 44s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/719f0100501c4bb19265692959adfb60 : SUCCESS in 22m 18s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3be47da862dd4d469a769ddaa7e98ff6 : SUCCESS in 18m 30s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a00b900b46e44d29bd32bc1872f2084 : SUCCESS in 19m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ca95993312344dd9ada54c2db2932d9 : SUCCESS in 19m 48s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c734247002a49c195ce9944df148f4b : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c6b3c4349a5672aebc47725eda32f17fe44ebc11","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 09:20:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a62fdeba7e1f82f7552bb621192e458e5a0ca47c","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-25 11:04:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"66a75e1d08634480bd6c76f8901051e5c21891cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:04:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"52860a46482e125f27a19c20285f64fa908a693b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:23:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3b0ddd5054764baa9e256a95a1fc87a1 : SUCCESS in 6m 15s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cd4f008dedd046ad8353f347c67f236e : SUCCESS in 9m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8a16852b531b4a03bfd2963a00ae35ef : SUCCESS in 12m 21s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfd908ce8ea24a3db3fb83e7deb8253b : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acad2455c26941bf84285b2c0efccf37 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":4},{"id":"3014cc04fff16fb9bd6a852d5f63548369aa13da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:23:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"445c73c91cbf86e6362c8bc2007d0bf74fad88f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:26:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe46eb1cc32e4993bb5e7f75e0bac13c : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1ba90cfd3a050c4147421a959b7f2ea3fc86ca9b","revisions":{"31659e97bd1c4144fb865d564c7ef0840043c61b":{"kind":"REWORK","_number":1,"created":"2020-06-24 13:05:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/70/737770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/737770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/1"}}},"commit":{"parents":[{"commit":"2171a3b529236455a4ead95ec55b183f2b335eec","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2171a3b529236455a4ead95ec55b183f2b335eec"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 15:49:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-24 13:05:57.000000000","tz":0},"subject":"Add functional tests for ubuntu focal","message":"Add functional tests for ubuntu focal\n\nDepends-On: https://review.opendev.org/732010\nDepends-On: https://review.opendev.org/732013\nChange-Id: Ifcb09830dc6d4c9f06db907a4b646a230b1ad938\n(cherry picked from commit 123cdc3bfe2d1ad2684136039c99aa3cccae1b6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/31659e97bd1c4144fb865d564c7ef0840043c61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/31659e97bd1c4144fb865d564c7ef0840043c61b"}]},"branch":"refs/heads/stable/ussuri"},"fd05167ba8fcc87f8c32518c337b621427123ef6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-24 13:07:54.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/70/737770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/737770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/2"}}},"commit":{"parents":[{"commit":"2171a3b529236455a4ead95ec55b183f2b335eec","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2171a3b529236455a4ead95ec55b183f2b335eec"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 15:49:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-24 13:07:46.000000000","tz":0},"subject":"Add functional tests for ubuntu focal","message":"Add functional tests for ubuntu focal\n\nChange-Id: Ifcb09830dc6d4c9f06db907a4b646a230b1ad938\n(cherry picked from commit 123cdc3bfe2d1ad2684136039c99aa3cccae1b6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd05167ba8fcc87f8c32518c337b621427123ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd05167ba8fcc87f8c32518c337b621427123ef6"}]},"branch":"refs/heads/stable/ussuri"},"dc0e932976684f3b8345035b992b762996031dd5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-25 06:49:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/70/737770/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/737770/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/3"}}},"commit":{"parents":[{"commit":"efbc473b01dcc7cc4b265132f93437e479ea0cdc","subject":"Stop installing python2.7 on debian/focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/efbc473b01dcc7cc4b265132f93437e479ea0cdc"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 15:49:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-25 06:49:33.000000000","tz":0},"subject":"Add functional tests for ubuntu focal","message":"Add functional tests for ubuntu focal\n\nChange-Id: Ifcb09830dc6d4c9f06db907a4b646a230b1ad938\n(cherry picked from commit 123cdc3bfe2d1ad2684136039c99aa3cccae1b6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/dc0e932976684f3b8345035b992b762996031dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/dc0e932976684f3b8345035b992b762996031dd5"}]},"branch":"refs/heads/stable/ussuri"},"1ba90cfd3a050c4147421a959b7f2ea3fc86ca9b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-25 07:42:40.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/70/737770/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/737770/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/737770/4"}}},"commit":{"parents":[{"commit":"efbc473b01dcc7cc4b265132f93437e479ea0cdc","subject":"Stop installing python2.7 on debian/focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/efbc473b01dcc7cc4b265132f93437e479ea0cdc"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 15:49:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-25 07:42:38.000000000","tz":0},"subject":"Add functional tests for ubuntu focal","message":"Add functional tests for ubuntu focal\n\nDepends-On: Iaf5fb24146b7203e879feccc7c96b0984b425d01\nDepends-On: Id21b3bc6721511d8d4a6b3e16ac465d14ec1a7f6\nDepends-On: I699ca3b6fbd35982c4fe384fbbdecf60e92e8d91\nChange-Id: Ifcb09830dc6d4c9f06db907a4b646a230b1ad938\n(cherry picked from commit 123cdc3bfe2d1ad2684136039c99aa3cccae1b6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1ba90cfd3a050c4147421a959b7f2ea3fc86ca9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1ba90cfd3a050c4147421a959b7f2ea3fc86ca9b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
