)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~737776","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~I583f2236acbc02b33e6fc192faa8cee62cd10b89","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","hashtags":[],"change_id":"I583f2236acbc02b33e6fc192faa8cee62cd10b89","subject":"Do not set PYTHONPATH for tox tests","status":"MERGED","created":"2020-06-24 13:23:33.000000000","updated":"2020-06-25 14:49:36.000000000","submitted":"2020-06-25 14:49:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737776-1593096576378-5ae3cf4f","meta_rev_id":"f0afdc5a16edd31d6f3e3063c1503e56a6e9e60b","_number":737776,"virtual_id_number":737776,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 14:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-06-25 13:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-24 14:46:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-06-25 13:32:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 13:32:33.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-25 14:49:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db54d0b00c581d355e56e546a9cf72d13e7e9927","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 13:23:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fc4035d4797c0699e6ed097af33efa28caef9920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 14:36:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b974a5b2c3a5426b9f4e28703b4bd000 : SUCCESS in 10m 00s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a45c97e3f0cc49278b56f06e54cefbac : SUCCESS in 10m 18s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/94a8f98661084061841a92cacf195c6e : SUCCESS in 13m 30s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a2185039e5ca48eea38a8c317d3a5152 : FAILURE in 8m 23s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45d4ccaad8744214b141c99bec51eacb : SUCCESS in 11m 33s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b635874e08ef44b9b6e32f5b3f2582e4 : FAILURE in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce6e381150b4a6e8ea68a0e9c5822fa : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5da60dda15ac4416bfd92ddb61f03ade : SUCCESS in 7m 15s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/a21a0d712d6c4f8c8b6f186473934a19 : SUCCESS in 18m 39s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5433b6b4c3d47ef9ef32e81a392d3a8 : SUCCESS in 16m 46s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e6a6c995f994474b8029a104ea5fbc4 : SUCCESS in 16m 58s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a887cbf659e4383abb1d5f889a4d6c1 : SUCCESS in 15m 15s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/cab743d23c6d4990a34673feba98073b : SUCCESS in 22m 11s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6122f04b366546988103f71a8fdfaeca : SUCCESS in 18m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af29c692763f47d48de98509c5ac8446 : SUCCESS in 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e8b25c8dfc072806127c2d9f13b681bae0b6100b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 14:46:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a7b540993a6b734ed71d292a6a18f1ebe6d9ac82","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-25 13:32:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f184118f88436142e106101a355d3fc0ac20809f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 13:32:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d2f03fb61243fa248a399fcd1beeef53035f8d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:49:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/de47760915cc4440b3fa7427c3ce5774 : SUCCESS in 7m 37s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0e3c771dbfd54d3aa9229e0bb0be09ab : SUCCESS in 10m 58s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3358267813714894ba143ccbacc1e6de : SUCCESS in 13m 01s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc24d6d9745144c5a7f1a19b5ac0dd65 : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f8f352106744393915994acb47bbc4a : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37df73c04e92448ebbf83bcba3e2b233 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f0afdc5a16edd31d6f3e3063c1503e56a6e9e60b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:49:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7789d476d4963734bc2b15a1b34c4734ee5b65a4","revisions":{"7789d476d4963734bc2b15a1b34c4734ee5b65a4":{"kind":"REWORK","_number":1,"created":"2020-06-24 13:23:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/76/737776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/76/737776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/737776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/737776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/737776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/737776/1"}}},"commit":{"parents":[{"commit":"2171a3b529236455a4ead95ec55b183f2b335eec","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2171a3b529236455a4ead95ec55b183f2b335eec"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-16 08:25:18.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-24 13:23:33.000000000","tz":0},"subject":"Do not set PYTHONPATH for tox tests","message":"Do not set PYTHONPATH for tox tests\n\nWe are now much clearer about the setting of ansible_python_interpreter\nand even though ansible is executing inside the venv it should\nbe deploying against the host with ansible, not the venv.\n\nDepends-On: https://review.opendev.org/735289\nChange-Id: I583f2236acbc02b33e6fc192faa8cee62cd10b89\n(cherry picked from commit 9468f9aad29bb7b37f03073a8c165bad371e433f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7789d476d4963734bc2b15a1b34c4734ee5b65a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7789d476d4963734bc2b15a1b34c4734ee5b65a4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
