)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~737979","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I0d82c76a35b0512a1366c55e06cde72f6b2ce3bd","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I0d82c76a35b0512a1366c55e06cde72f6b2ce3bd","subject":"Do not install python2 or its dev package on Ubuntu Bionic hosts","status":"MERGED","created":"2020-06-25 11:20:42.000000000","updated":"2020-06-30 13:04:47.000000000","submitted":"2020-06-30 13:00:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737979-1593522026355-fca765a5","meta_rev_id":"6d4bc38c55c13cd03c7d468e75bef834189e978d","_number":737979,"virtual_id_number":737979,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 13:00:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-30 12:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-29 07:47:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-06-30 12:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 14:06:14.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-30 12:37:31.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-30 13:00:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a2bcb12bce9476ce20edef9d1f12282e063f340","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 11:20:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2870bf246f28b62ae2605069b43151d18fc1d3f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:59:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f8b84cf8cef54601a6a9378b4703203d : SUCCESS in 9m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/54fceef3f80844e6bbf6e92a4a831b56 : SUCCESS in 10m 14s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7c08efeeb9a74dc9bfcf2ac9a5304a95 : SUCCESS in 12m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/fa3ae73d772e48299bf1a08073192c04 : FAILURE in 8m 51s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81f69b8a8e4e4c72ae0c6596b95add16 : SUCCESS in 9m 33s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53559c6621684a12a2c059ccf3960253 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94548774b151407da27587be43ff676e : SUCCESS in 5m 25s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/f87e4db0999f42418954d8193857ac03 : SUCCESS in 16m 14s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/468dda2722d94df7bc412f8e391376e9 : SUCCESS in 15m 27s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2243bc3100c441df90ecb8c4ea64b770 : SUCCESS in 16m 03s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/894435b7b4874e9c89b46bc4e8f9adc3 : SUCCESS in 17m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60d0f35d8b284f7ab9a275700c77ca9e : SUCCESS in 13m 49s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/531fe3935c9b4921956a4cdb5fc9a01c : SUCCESS in 11m 51s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fd97d8df5854fe1bbfb179341585779 : SUCCESS in 13m 29s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/37a4e7ce075f4a0280f6b1d04df01071 : SUCCESS in 16m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efb423d289204dd3a8d5a82786d4daea : SUCCESS in 16m 34s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e1e9c542dc4c5f83251e478201af4d : SUCCESS in 14m 51s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc136ac3f03d498d85be32c463e3e81b : SUCCESS in 14m 25s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0988d53634634b668afa5938e5af70d8 : SUCCESS in 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7635ded1135d0b137c8aadfb00703f04c4566975","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-25 14:06:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"282db9ba04af21f01d98498c18a1f201384a17fd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-29 07:47:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a778f767cf06cd9bee1932c5a5eb3abd4bd9431a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:20:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9c6786fcb4f140259db8d47f1fe54c69 : SUCCESS in 8m 26s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/479b635edba249fca67b8e0fc65cd0ba : SUCCESS in 9m 58s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ed2677128e3d4a2bba24750828a927d7 : SUCCESS in 10m 05s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c6cb8a7b80ef4f5c82df8b8cbcbc244d : FAILURE in 10m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3d3e48bb22342b3b8a4a110b1d7e5b6 : SUCCESS in 10m 11s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53a3ed631c3348319ed160e3479faade : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc712559d6f400faa215a5106b48c43 : SUCCESS in 6m 00s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/d7f70fa0c4c746c68994c89a96557723 : SUCCESS in 14m 39s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f9f7cceff1542729a003f5ad7a6b448 : SUCCESS in 16m 56s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aec3254f11d454c90676f2836976c7f : SUCCESS in 13m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0cb442523aa42eba446833d115ca3a0 : SUCCESS in 16m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3dd16782a5a4776ba262ed677c2160c : SUCCESS in 28m 12s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd6e4232d25843f48b1189ad1fe3855f : SUCCESS in 19m 29s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bfcb0b627144743aeab687dd1ddc36c : SUCCESS in 15m 07s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/c3648b7faeb34484ba81780f122b3498 : SUCCESS in 15m 35s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d313ae400b4d4f64ad032510f5a91deb : SUCCESS in 15m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86d7134ec9d8478fb4f74798671252ce : SUCCESS in 18m 22s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61ef0252a021441db15551aad419293f : SUCCESS in 16m 36s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8e25bee0ccd4e9cb5bd1ca51e4f1a32 : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":2},{"id":"82fb0bfd03fd3e89dac8885ac81e2d5fa924b1ea","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-30 12:37:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ce2a2268824fa8276e2c20bca1735912323964d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 12:38:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"585df7e8dc4bcab9d172eba725b5eba1cf025b0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:00:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"448425b2b02d80957ce36badbde5e754e375473e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:00:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/20360410f7a64753801c7b8c5aa40e01 : SUCCESS in 9m 40s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1a38c37f0aa94d6084c89de993ae0843 : SUCCESS in 9m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/20b15d164ec8439fb2dc26e1c5872fb0 : SUCCESS in 13m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff3406b41c734935be9cb761c270c434 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618f66369e3146a79d544521910bf5a1 : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9b731c806dd82726f051aa343170baa026570842","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-30 13:02:16.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 31ceae627e161b463ec20a055988d8f8e3ebaca2","accounts_in_message":[],"_revision_number":2},{"id":"6d4bc38c55c13cd03c7d468e75bef834189e978d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:04:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b700b1a011e409a843335f90debc884 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f015e6fed9adf44b96d8d6ccaaceaa881ab0295","revisions":{"48ec9cbd7cd82eac952680f91256904348e6d75e":{"kind":"REWORK","_number":1,"created":"2020-06-25 11:20:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/79/737979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/79/737979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/79/737979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/79/737979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/79/737979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/79/737979/1"}}},"commit":{"parents":[{"commit":"a59f9d8ec71cf82e38a8ea00d446c7e656f23ac5","subject":"Remove support for Unubut 16.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a59f9d8ec71cf82e38a8ea00d446c7e656f23ac5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-25 11:19:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-25 11:19:21.000000000","tz":60},"subject":"Do not install python2 or its dev package on Ubuntu Bionic hosts","message":"Do not install python2 or its dev package on Ubuntu Bionic hosts\n\nThe CI images are now the \u0027plain\u0027 variants so we do not need python2\nto be installed on the host\n\nChange-Id: I0d82c76a35b0512a1366c55e06cde72f6b2ce3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/48ec9cbd7cd82eac952680f91256904348e6d75e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/48ec9cbd7cd82eac952680f91256904348e6d75e"}]},"branch":"refs/heads/master"},"3f015e6fed9adf44b96d8d6ccaaceaa881ab0295":{"kind":"NO_CHANGE","_number":2,"created":"2020-06-29 07:47:23.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/79/737979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/79/737979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/79/737979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/79/737979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/79/737979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/79/737979/2"}}},"commit":{"parents":[{"commit":"ac2789cebd75d741f099b3328c42278f81b4a7df","subject":"Remove support for Ubuntu 16.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ac2789cebd75d741f099b3328c42278f81b4a7df"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-25 11:19:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-29 07:47:23.000000000","tz":0},"subject":"Do not install python2 or its dev package on Ubuntu Bionic hosts","message":"Do not install python2 or its dev package on Ubuntu Bionic hosts\n\nThe CI images are now the \u0027plain\u0027 variants so we do not need python2\nto be installed on the host\n\nChange-Id: I0d82c76a35b0512a1366c55e06cde72f6b2ce3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3f015e6fed9adf44b96d8d6ccaaceaa881ab0295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3f015e6fed9adf44b96d8d6ccaaceaa881ab0295"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
