)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~738608","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~I758360886f77ffc20856bc00424a3c7027c564f3","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","hashtags":[],"change_id":"I758360886f77ffc20856bc00424a3c7027c564f3","subject":"Remove support for Ubuntu 16.04","status":"MERGED","created":"2020-06-30 13:05:01.000000000","updated":"2020-07-06 12:39:28.000000000","submitted":"2020-07-06 12:37:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738608-1594039060308-506a25db","meta_rev_id":"c511ea8e0309122e66f8b0ff67846238c61b5acf","_number":738608,"virtual_id_number":738608,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 12:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-07-06 12:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-05 19:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-07-06 12:16:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 12:16:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-07-06 12:37:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ca504e622e2fcae322c606b737331d72f8ce164","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-30 13:05:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8b86eaa0c85926c18d3fcd2e181b0b2e9e4ce31d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 13:38:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/828ccfb2169945a2bf6590b74c887179 : SUCCESS in 7m 13s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6cb2bfa678ce42a7a40020ce40544917 : SUCCESS in 10m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0affa0232196495bb0c7ab0008c3bfdf : FAILURE in 10m 19s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/e6386500e15c41fd980581d354d0a005 : FAILURE in 10m 14s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecbc3128f63842fcbd7748c426c9a238 : SUCCESS in 12m 46s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9071d7916ba34c6e836d5404aa25db3e : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1645726152bd46fdba54a430e9b268ea : SUCCESS in 7m 55s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/44aaa24d1d734c11b43eb348d21ef7eb : SUCCESS in 30m 14s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/def6a3e1fb23425bab91422e19ad1c0c : SUCCESS in 18m 44s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aff9be471d3b489bac1be7dadab5daea : SUCCESS in 19m 53s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e034425471c464c997024d8a6bd5abd : SUCCESS in 20m 46s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66d277b8899c4dd6bf5d0d30103b6881 : SUCCESS in 15m 27s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/405c11a6f83e49febb80ac98735848dc : SUCCESS in 13m 22s\n- openstack-ansible-lxc-machinectl-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d299b796d97f4a81827b48f3e29e0183 : SUCCESS in 14m 13s\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/e0750815a13a46a5aff9b4a5fa87011a : FAILURE in 11m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8787dd1130b42c2b8d09e06a03fc314 : SUCCESS in 13m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d24c53d0d3a64398a6af82eccd70ae40 : SUCCESS in 13m 50s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29648863e92542e680f367875023d3e4 : SUCCESS in 13m 50s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ea6c8ab89354a5885014b67590abcb7 : SUCCESS in 16m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1bf6ec0e12c7ac9bcbedc24db25526991a62dc6b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-05 17:47:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45b371db29f724eb4c08caf04137c5caf48ae8b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 18:15:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a90c1ef1cad24453a30a7c3e8d6ef4c9 : SUCCESS in 5m 49s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/36af516fd9c240828dd527d7435db360 : SUCCESS in 7m 32s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4f9c7760b23743e89a50b9cea1e93e4e : SUCCESS in 8m 34s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/aa8f5ae87a474af18f3467339cfddfc9 : FAILURE in 6m 42s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/387fa2c942f64152905579a2854fd170 : SUCCESS in 7m 32s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09606c5dc1094a4eae02f938a8e5501b : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab610db914e439daaa224207867524a : SUCCESS in 4m 51s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/63fb0c54e90b44259d68cd3c3063bfe7 : SUCCESS in 12m 09s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe84b55e622c4bf9b500da620f371932 : SUCCESS in 11m 18s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9ad6bd5969d4b2d82a0bc6f6e841303 : SUCCESS in 11m 44s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6da378251b494f8b84e8673b82bbd771 : SUCCESS in 11m 59s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d940395286e94813874e667c6f5f260e : SUCCESS in 12m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/626830ab3965470282d34ecbbf1db6a9 : SUCCESS in 11m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f3229c2c1b3465ba1fd182acfa4c2e3 : SUCCESS in 12m 05s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b40d6ac1b1a74defa71f4067c79e31d9 : SUCCESS in 11m 52s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5c073a81fea46cb813b06f79630eb68 : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f3215157911dc1f30905d302b03444e1b46cd517","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-05 19:59:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"91c5c98b50e296ffc547de64ac6247513d0c1c78","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-06 12:16:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"458c246b3aa6713c3e811174d5c6a9cb6b256e75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 12:16:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4ec44eb6b0e71d29ab9de76a43e07c473135f54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 12:37:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2e80aeb8d80c556b84aa40ca39067c95fff3697b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 12:37:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b009ade0bd1f4945b431dbdb03bcd7e0 : SUCCESS in 7m 17s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/57ed8edf695a4ca1a026ecb7fd14d0cb : SUCCESS in 9m 16s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/90a902bfbc7441d58efd361d853a6160 : SUCCESS in 13m 07s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c5fb1edcc16483e8c75b642108f2a1e : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ae6d0527b24f449f1431d391b7cce8 : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c511ea8e0309122e66f8b0ff67846238c61b5acf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 12:39:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fad7905c57445f297e121550ef7bff0 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16ea6704f788772851d74cad6f65631d88cac2e6","revisions":{"16ea6704f788772851d74cad6f65631d88cac2e6":{"kind":"REWORK","_number":1,"created":"2020-06-30 13:05:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/08/738608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/738608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/738608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/738608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/738608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/738608/1"}}},"commit":{"parents":[{"commit":"ddf83135219acf6b6da0029b8a4383aa3d79857e","subject":"Merge \"copy the actual keyring\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ddf83135219acf6b6da0029b8a4383aa3d79857e"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-25 11:18:53.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-30 13:05:01.000000000","tz":0},"subject":"Remove support for Ubuntu 16.04","message":"Remove support for Ubuntu 16.04\n\nWe no longer support Xenial for openstack-ansible since Rocky.\n\nChange-Id: I758360886f77ffc20856bc00424a3c7027c564f3\n(cherry picked from commit ac2789cebd75d741f099b3328c42278f81b4a7df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/16ea6704f788772851d74cad6f65631d88cac2e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/16ea6704f788772851d74cad6f65631d88cac2e6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
