)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~742124","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I395e0f7b1d362240e67a86fa4545a8be64f3053c","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-remove-centos7","hashtags":[],"change_id":"I395e0f7b1d362240e67a86fa4545a8be64f3053c","subject":"Remove support for Centos-7","status":"MERGED","created":"2020-07-21 11:19:15.000000000","updated":"2020-09-28 10:42:49.000000000","submitted":"2020-09-28 10:37:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742124-1601289432268-1561717f","meta_rev_id":"29f40f2f366c29d657a20e422cc14949f344a199","_number":742124,"virtual_id_number":742124,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-28 10:37:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-09-28 10:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-09-25 09:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"date":"2020-09-22 13:27:51.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-09-28 10:20:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 13:27:51.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-09-25 09:04:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-09-28 10:20:03.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-09-28 10:37:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1c460eee03deb5c3d4840220f1ab70dd7ed4c77","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 11:19:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c8b694ee8acfea8df7b47bb26b2ada9098f61be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 11:44:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/42c3b96b141d4897abec873f39841596 : SUCCESS in 7m 50s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/186a3ffe398544f0ac45a605bf127e4c : SUCCESS in 8m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9bcf9151c0284aed8381b3a2849339ff : FAILURE in 9m 47s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a35a94d8cde24eb5b6676ea7c01e6835 : SUCCESS in 13m 23s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b3f0438eed94292b8c5ddda85322d4b : SUCCESS in 9m 47s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dae6e46f16446a0a1352e5d76623508 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920b18ac0de94bd9ad1673a0b0a43417 : SUCCESS in 5m 39s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/437eab4e9256458b815f9083c2a9eb2c : FAILURE in 9m 24s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1790b538388c453a9eaef1f7a0828a7b : SUCCESS in 15m 49s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f888932af7a74982852cb55a903ee1e3 : SUCCESS in 17m 44s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fbdaf9052624582a64a9b638bb22cf1 : SUCCESS in 15m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91621428eef84c6f92397887d700de6c : SUCCESS in 15m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bcfc6bb7d3b4deca1a1e188de52b770 : SUCCESS in 15m 17s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17a10d2ee2ed4838ba63b905777d06bd : SUCCESS in 14m 24s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f381d80fa5d4dd989d8324338129f21 : SUCCESS in 14m 46s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5902cdeee05e482a91dd96fa2f344026 : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a0d2bbc4a6eb9378b0473570fd1cebac758c08d8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 12:39:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6d52fb758ae297ff1325534a4fd950e9bb80293a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 13:07:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b545e4e1cd0d4605b02b85f3c442c1ac : SUCCESS in 11m 33s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ca1b277d94644815b5ec7a55fb742cca : SUCCESS in 12m 43s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/52029f2a52ee499ab5b4ca954df9b718 : SUCCESS in 14m 01s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed2487fc3542453aac4f7a7213908fd6 : SUCCESS in 11m 16s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2acc4e46348a497b9ac6563558774264 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d13bebca18485da23136383c618e97 : SUCCESS in 6m 18s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/9bf1fbbb3e3b4129be2521697368aff2 : SUCCESS in 16m 36s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3f2e77626e7485892eca08d62e0df29 : SUCCESS in 13m 23s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94d8f5c3202b4f10b92efe8ed9a85c4e : SUCCESS in 15m 09s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/deaae215e7b642eb9ec47ec8bd1719e8 : SUCCESS in 16m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f3148ff909e4695add9ff084a5ba2a4 : SUCCESS in 15m 43s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d794e456d40a42f780f3472daf9de596 : SUCCESS in 15m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbff4ca7779b4c778dd1568493eeceed : SUCCESS in 15m 07s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee504832f4764cb5a37ae90aae81758e : SUCCESS in 17m 43s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c3ae5d7a7f443528e02b8c471e3278c : SUCCESS in 17m 25s","accounts_in_message":[],"_revision_number":2},{"id":"cdef6020cb5238fc0050030c83b675ef58165071","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-14 12:42:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5242e75b8d9671d865f5c5218bd31249e18bf640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 13:19:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7f504e13d98b41babcb68d74fa9f7c2b : SUCCESS in 5m 56s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0c00e2a7eff24119b71805c9c6cd5306 : SUCCESS in 9m 27s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/21e8136e96df4a85a445912a781c5daa : SUCCESS in 15m 08s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a04609d3636b48779acc2ab61decbe13 : SUCCESS in 11m 03s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/668e4772001e4a6db2998300f05250f4 : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff58d1c78694027a302cd46ef337da0 : SUCCESS in 12m 00s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/37e2ee140447461f9553e93b7fc1b3d2 : SUCCESS in 20m 04s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1e406ee460e462e89b52664c0bf1b9a : SUCCESS in 13m 31s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93fdc7f743d4481695795b5f9905a5d1 : SUCCESS in 17m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d4fc54dad704f818900575a0e3f77ad : SUCCESS in 18m 15s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9070148d9aa5415788ded09bc67f9793 : SUCCESS in 12m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7eb36fbbc9744d1a89d3ba2bc68bb2a : SUCCESS in 14m 16s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ecaf3fe6c7b4fd59cffaee62ecb34de : SUCCESS in 16m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fccea3172c69485fbddb773d97c8db41 : SUCCESS in 16m 54s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/385408f9e7274d0db2ca8da7ac906e6a : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a87fe4826dac25a80bbd046a3f27c1ed089973cc","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-09-22 12:53:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5af912bb8842a10cc0d88242d1fc14c86b7f86a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 13:19:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b2c289ea302a48e38c8dce858aa3daf2 : SUCCESS in 6m 49s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7cf85ff46b8f476bb51eea7944d098c6 : SUCCESS in 8m 54s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/dbd2edbbfa584a36a1ad75694724ef4b : FAILURE in 6m 36s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e530387a6f4b415f9e2b5ef31634da8f : SUCCESS in 9m 32s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2adcdb3563c24f88b065da19183f66e7 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdbd7eeeff1b4b0489e7a4e5a50d89e5 : SUCCESS in 11m 22s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/8f3d37c1e1f14dc1805d6f2c7eeab64e : SUCCESS in 18m 00s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2de920b2159644c4817043a3b6c05147 : SUCCESS in 15m 25s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/147e7cb60b6a401c9b5c49a26c2021f2 : SUCCESS in 16m 25s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/758f21518fb34206b7eb81db4795f61e : SUCCESS in 16m 36s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/543f43a5b0da463ab648334f386b3487 : SUCCESS in 16m 03s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6056f158d8046f0af48bc396bb8ba03 : SUCCESS in 14m 15s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31b54998f0f344c18f3134072bc362cd : SUCCESS in 14m 08s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f5938e0117840a193ac2ed33d9129d7 : SUCCESS in 15m 09s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/283839b1c7734b9f840dfcd649cb9b9e : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":4},{"id":"0591951583e5ea0edef9d4798929926911a24e6d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-09-22 13:27:51.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7b9fd5e2282d1efb54dfd601ca6cd9cd22ebe55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 13:50:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8ab9cb78f519469b961d40871d6a7c04 : SUCCESS in 6m 47s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ea31ea2102ac431abf4a1291e52e29b4 : SUCCESS in 8m 29s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4443c7a7845246acab17e838ea24464f : SUCCESS in 11m 23s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a24ddeeec0fa40aeb6d1497e095f3f5c : SUCCESS in 9m 35s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e38b619a6564ba992046cbb629ba6f0 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e3a578bfa848489f0d120c146cda36 : SUCCESS in 12m 01s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/b678c04da65f4ebdb0537452a1131f44 : SUCCESS in 18m 15s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99edfc22508e4b9988ec04d496ee4342 : SUCCESS in 15m 00s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c6a938290cf4d70966fca14286eead7 : SUCCESS in 15m 02s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c02cc5ee4864501a26d6d2bdf8c4565 : SUCCESS in 16m 09s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eb44b44be444a24813a9cb060cabca2 : SUCCESS in 13m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75598f5e0e04464bb6536dc57d4050b1 : SUCCESS in 15m 12s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdfffc4f72e5436388ba5ecefa8c9e7f : SUCCESS in 16m 04s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/809eba547f2b49159eef80ca2fdf0b45 : SUCCESS in 16m 04s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbaf244b87994d9f96ef218534584936 : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":4},{"id":"c90f1ab415645d97e8bbe2d95e33dc3527b68fee","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-24 09:11:41.000000000","message":"Topic set to osa-remove-centos7","accounts_in_message":[],"_revision_number":4},{"id":"64c8e70eb4e7feb3785e6726e6a13c2bea644cba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-25 09:04:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0be9256f6d2d74e763d3eb57ff6290141f697edd","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-09-28 10:20:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bc5cc36769df1de89e41ace92dca2a748e216181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 10:20:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"51edeaa415a58e85a9654f7e1af579c4417b11b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 10:37:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ebc4fb144e0f1944d46a09c92a6a8a79c4d5419c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 10:37:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/344e13b24f85487b86bc6f2b1b3bf08e : SUCCESS in 5m 51s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c89e7961c7c34c05addbf0502bef7be4 : SUCCESS in 8m 55s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/449c48c12d164e68867ea611ced1ae3d : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad10013852bd42a295fdcf8fc33afae2 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":4},{"id":"29f40f2f366c29d657a20e422cc14949f344a199","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 10:42:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4b653ca77e42b7a30a66582bea09c5 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cfcaff62e8f34f2c4aab572d01600d839a5fd0ed","revisions":{"ee3684ae4b5179b1012ca8d9bba3a426cd54a70a":{"kind":"REWORK","_number":1,"created":"2020-07-21 11:19:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/24/742124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/24/742124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/1"}}},"commit":{"parents":[{"commit":"eaae389a4bf9708eb32bb820429da6343edeb7d5","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eaae389a4bf9708eb32bb820429da6343edeb7d5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 11:17:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 11:19:00.000000000","tz":60},"subject":"Remove support for Centos-7","message":"Remove support for Centos-7\n\nOpenstack-ansible does not support Centos-7 beyond Ussuri so drop\nsupport for Victoria.\n\nChange-Id: I395e0f7b1d362240e67a86fa4545a8be64f3053c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ee3684ae4b5179b1012ca8d9bba3a426cd54a70a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ee3684ae4b5179b1012ca8d9bba3a426cd54a70a"}]},"branch":"refs/heads/master"},"67d8866486c13602dfd4b5a81ea07d486a48b8d7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-21 12:39:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/24/742124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/24/742124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/2"}}},"commit":{"parents":[{"commit":"eaae389a4bf9708eb32bb820429da6343edeb7d5","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eaae389a4bf9708eb32bb820429da6343edeb7d5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 11:17:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 12:39:12.000000000","tz":0},"subject":"Remove support for Centos-7","message":"Remove support for Centos-7\n\nOpenstack-ansible does not support Centos-7 beyond Ussuri so drop\nsupport for Victoria.\n\nDepends-On: https://review.opendev.org/742103\nChange-Id: I395e0f7b1d362240e67a86fa4545a8be64f3053c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/67d8866486c13602dfd4b5a81ea07d486a48b8d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/67d8866486c13602dfd4b5a81ea07d486a48b8d7"}]},"branch":"refs/heads/master"},"9c18f87941ae23755b7a9f7cbe3605fdcabd3f2b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-14 12:42:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/24/742124/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/24/742124/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/3"}}},"commit":{"parents":[{"commit":"c21fcdb58380c3723b4dccb9c91824c2ef4ce159","subject":"Merge \"move listchanges removal to where it\u0027s needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c21fcdb58380c3723b4dccb9c91824c2ef4ce159"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 11:17:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-14 12:42:19.000000000","tz":0},"subject":"Remove support for Centos-7","message":"Remove support for Centos-7\n\nOpenstack-ansible does not support Centos-7 beyond Ussuri so drop\nsupport for Victoria.\n\nDepends-On: https://review.opendev.org/742103\nChange-Id: I395e0f7b1d362240e67a86fa4545a8be64f3053c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9c18f87941ae23755b7a9f7cbe3605fdcabd3f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9c18f87941ae23755b7a9f7cbe3605fdcabd3f2b"}]},"branch":"refs/heads/master"},"cfcaff62e8f34f2c4aab572d01600d839a5fd0ed":{"kind":"REWORK","_number":4,"created":"2020-09-22 12:53:11.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/24/742124/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/24/742124/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/24/742124/4"}}},"commit":{"parents":[{"commit":"c21fcdb58380c3723b4dccb9c91824c2ef4ce159","subject":"Merge \"move listchanges removal to where it\u0027s needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c21fcdb58380c3723b4dccb9c91824c2ef4ce159"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 11:17:05.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-09-22 12:53:09.000000000","tz":60},"subject":"Remove support for Centos-7","message":"Remove support for Centos-7\n\nOpenstack-ansible does not support Centos-7 beyond Ussuri so drop\nsupport for Victoria.\n\nDepends-On: https://review.opendev.org/742103\nChange-Id: I395e0f7b1d362240e67a86fa4545a8be64f3053c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cfcaff62e8f34f2c4aab572d01600d839a5fd0ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cfcaff62e8f34f2c4aab572d01600d839a5fd0ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
