)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~742698","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I70f6de497fa3729a1d19540fad03e273f260433d","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"wsoyinka-openstack-ansible-lxc_hosts-patch1","hashtags":[],"change_id":"I70f6de497fa3729a1d19540fad03e273f260433d","subject":"Add remote_src parameter to ansible unarchive module for lxc_hosts task","status":"MERGED","created":"2020-07-23 15:56:59.000000000","updated":"2020-07-24 08:20:19.000000000","submitted":"2020-07-24 08:18:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742698-1595578693996-da4fe202","meta_rev_id":"f7830af9cf3d11b6013ec62076590e14bb32cb33","_number":742698,"virtual_id_number":742698,"owner":{"_account_id":32197,"name":"ws","email":"wsoyinka@gmail.com","username":"soyinka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 08:18:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-07-24 07:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-24 07:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-07-24 07:57:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 07:02:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-24 07:57:38.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-07-24 08:18:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0b923ae46ddc59ae1cc2e79739a0345d34359f8","author":{"_account_id":32197,"name":"ws","email":"wsoyinka@gmail.com","username":"soyinka"},"date":"2020-07-23 15:56:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcac1425084493a5c5e65c6d154e952a8a8309d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 16:30:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/35e558d1505e420586876ec7babc8947 : SUCCESS in 9m 19s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/79c9aecd7c34438bbe4e06e6411f7933 : SUCCESS in 12m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a46539049fbe4145b1d3c4a47986a19e : SUCCESS in 11m 18s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/0f37e74d9a254100bc655c8cd0a33177 : SUCCESS in 14m 16s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8546501a9acf45efa09ca29579712b3e : SUCCESS in 13m 03s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1acd797bd7ff4b3db2450e15dc31497b : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d782541fd05b4e268b89fe5d87c09b2c : SUCCESS in 8m 37s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/8ec763c2e9d24aadb4646a3f65066f8f : SUCCESS in 19m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b160f1a56f04193b818e3c855fcee0e : SUCCESS in 21m 05s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2564e46c3a4446ed8fe5887517328fad : SUCCESS in 21m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edca7345818b41f9889520e42d89d1de : SUCCESS in 16m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b182f991a91747fb97232f2940c692f7 : SUCCESS in 18m 58s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe7a1deff4a64036a78cde078f0f8d06 : SUCCESS in 17m 17s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ee65cbcef174a60b555078e43442890 : SUCCESS in 16m 05s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73139d9eb7fd4ca7bfd376efa2a8fa78 : SUCCESS in 14m 17s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7103fa6ea0c4cf8852d0e76a283b405 : SUCCESS in 16m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f5d75f4004772e497304c4631ad08a109d168b5a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 07:02:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"645966901d007ae7979afa09c1ae5ff2af46a56e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-24 07:57:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc297b752fbc2c5dc69408ef967676cd4f480f2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 07:57:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12b727d633f7d8bf298ad44bb13ca1a841e4a11a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 08:18:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc44891db5d517cf34af4c8e9863184eaf9f7602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 08:18:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2174e4f8c8344e908e3b29845296f968 : SUCCESS in 8m 45s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b9fc62d64e10472d8f28820fa7e8fd9c : SUCCESS in 12m 07s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/85dd4f9c162a48788158b48b7fd27b06 : SUCCESS in 13m 06s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0735ba55ae54487899c7a79393e89ed4 : SUCCESS in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b467234b6a4c589bdbd120d52a186b : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6de44cf725d63f5464c8b8cd2c23556241f5bff7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 08:19:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f1424118cb45b09e74dbb345ca49f9 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7c782e36734650a9cb84ff51d3a6cfb02d9db0d0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 08:20:19.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 102b7780e85a354dd8ceff481d31519b1ff7bbba","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c94eccc383d281bec000980b1f8d4437a2c81558","revisions":{"c94eccc383d281bec000980b1f8d4437a2c81558":{"kind":"REWORK","_number":1,"created":"2020-07-23 15:56:59.000000000","uploader":{"_account_id":32197,"name":"ws","email":"wsoyinka@gmail.com","username":"soyinka"},"ref":"refs/changes/98/742698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/98/742698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/742698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/742698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/742698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/742698/1"}}},"commit":{"parents":[{"commit":"eaae389a4bf9708eb32bb820429da6343edeb7d5","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eaae389a4bf9708eb32bb820429da6343edeb7d5"}]}],"author":{"name":"wsoyinka","email":"wsoyinka@gmail.com","date":"2020-07-23 15:37:54.000000000","tz":-240},"committer":{"name":"wsoyinka","email":"wsoyinka@gmail.com","date":"2020-07-23 15:56:52.000000000","tz":-240},"subject":"Add remote_src parameter to ansible unarchive module for lxc_hosts task","message":"Add remote_src parameter to ansible unarchive module for lxc_hosts task\n\nlxc_hosts - Unpack base image tasks fails because ansible is\nlooking for CentOS-8-Container*.tar.xz image archive on\nansible controller instead of on remote host. remote_src\nparam fixes and lets job continue.\n\nChange-Id: I70f6de497fa3729a1d19540fad03e273f260433d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c94eccc383d281bec000980b1f8d4437a2c81558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c94eccc383d281bec000980b1f8d4437a2c81558"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
