)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~752905","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~I1d7dc2ff6bc1fef6d1d2e5e56b10012c7d575f5e","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","topic":"bug/1895518-stable/ussuri","hashtags":[],"change_id":"I1d7dc2ff6bc1fef6d1d2e5e56b10012c7d575f5e","subject":"Wait for aria2c to finish","status":"MERGED","created":"2020-09-21 09:38:02.000000000","updated":"2020-09-21 12:09:08.000000000","submitted":"2020-09-21 12:02:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752905-1600689727804-57e98530","meta_rev_id":"9a1e86fdb6b50eb303d793377dba2e97879ab397","_number":752905,"virtual_id_number":752905,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 12:02:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-09-21 11:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-21 11:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-09-21 11:37:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 11:24:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-21 11:37:48.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-09-21 12:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12ff538bcdf0640c1676bb8626b55408d51534c0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-21 09:38:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d22cafa9e7f5bb44ed2b5bfb3d3de94e68a317f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:02:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ccf04c94ba664ea59c35db045fe59349 : SUCCESS in 7m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b6f3b11692874d1fba2f71a1abdb8e78 : SUCCESS in 9m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1108a6d5593c49c78f0b4217aa01b907 : SUCCESS in 11m 30s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7519af43fefb42be87f1bbe6ce8e861f : SUCCESS in 11m 05s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0436081f12664014af1f25223a1128ba : SUCCESS in 10m 03s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d91547f16ca645f791d51cbd09b04656 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64505315c9ef4b72b3809d55da35811a : SUCCESS in 14m 40s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/cd0b268ffcf6453eaaf26096b8061fea : SUCCESS in 18m 20s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b902250409fa462cb23a7abd18bfe162 : SUCCESS in 14m 29s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71738c55fe0e406baa298e578a312ba3 : SUCCESS in 12m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50d94da5abd347e6af0389f15b2f62df : SUCCESS in 14m 49s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/460ff69d101e48c1937c8a625bf6050a : SUCCESS in 15m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5150775d2b26487b89b2897e85c1bcc9 : SUCCESS in 12m 57s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68bda95a12124525bb895711f15c9628 : SUCCESS in 14m 00s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65880dd2ee5d4ad38ef8b57454e0f2e2 : SUCCESS in 13m 37s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df8caba70e7d480b9d8682a07c4e96cc : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e4af3b46e9e8dbd52d762b01277dd7cabebc5e1b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-21 11:24:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c055a23cc66c7d8d271ce6ffbadf9af60f0245a6","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-09-21 11:37:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fefc9e96d8bf6d875df18789420f03521b0f7b2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 11:37:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"977c71efabf87a5065696f01597eecf5502b8c4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:02:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"25520cf04ddc6605579bc96e550a9630d0ef70be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:02:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1517dad998664668a6360e9e18d8e2ba : SUCCESS in 6m 10s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/41b45eb280d5492aa7db78a64ac37706 : SUCCESS in 13m 30s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4988fdadcc6447b890366d27d989ddd3 : SUCCESS in 13m 11s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd10acc2a3b3462b80ba73232b28ce33 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f149311e984edf8a30ed9a55f17562 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a091f2bed0873837bf401166f93ead419f9323e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:09:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318fb43a91ff4c3f8de9a725b1829da1 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c1a2183319a5ab46a04f1bc1baaf25fc1e2309f","revisions":{"6c1a2183319a5ab46a04f1bc1baaf25fc1e2309f":{"kind":"REWORK","_number":1,"created":"2020-09-21 09:38:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/752905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/05/752905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/05/752905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/05/752905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/05/752905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/05/752905/1"}}},"commit":{"parents":[{"commit":"cb4f10bdf1692be78790e06b21a3f60f7d337f8c","subject":"move listchanges removal to where it\u0027s needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cb4f10bdf1692be78790e06b21a3f60f7d337f8c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-14 10:07:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-21 09:38:02.000000000","tz":0},"subject":"Wait for aria2c to finish","message":"Wait for aria2c to finish\n\nWe launch aria2c in async without wait and hope that it will finish\nin time before we start unarchive. However it\u0027s worth checking that\nimage download has really finished instead of making such assumption.\n\nAdditionally ensure that ubuntu has gzip installed, as their image is\ngunzip compressed.\n\nChange-Id: I1d7dc2ff6bc1fef6d1d2e5e56b10012c7d575f5e\nCloses-Bug: 1895518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6c1a2183319a5ab46a04f1bc1baaf25fc1e2309f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6c1a2183319a5ab46a04f1bc1baaf25fc1e2309f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
