)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~754155","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I38f31860408ad878ad57aaacae7e7776fec57dc5","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I38f31860408ad878ad57aaacae7e7776fec57dc5","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-09-24 16:54:02.000000000","updated":"2020-10-06 06:47:55.000000000","submitted":"2020-10-06 06:45:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754155-1601966758859-d979b326","meta_rev_id":"ae825dd9e6ee0ab5ca894d945e634cf3e410a48a","_number":754155,"virtual_id_number":754155,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 06:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-06 06:29:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-06 06:29:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 06:29:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-06 06:45:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdb4bc9dc8cee17eb533fede3838e986c24d32e1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-24 16:54:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c3f4d9c3d42866f7f1b3aec49737ac82cfb7dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 17:22:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9de3730869774108a829e00418d8ee28 : SUCCESS in 6m 37s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6488ce01f91244258ef56ed4945aabcd : SUCCESS in 11m 34s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/572f082739e5488bb51906df46479950 : SUCCESS in 12m 44s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf5746739c264b34b15384974b7f09e3 : SUCCESS in 11m 11s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b81862f289ee4534862796bd9b65e5c8 : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee3b4d66d154b9db3626fd040576f88 : SUCCESS in 5m 55s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/b3b1b704d2e148c4a1af6346139577e4 : FAILURE in 4m 25s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83b026a623c9472fb4b8e7e47c6c783a : SUCCESS in 13m 18s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d9c4ebb9a6f4856b6288ac4e021b07d : SUCCESS in 15m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a49fe46bd7c4e2aaf2ebe769fbed098 : SUCCESS in 15m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f49e5ea9359d4f33991cf16c12ed035f : SUCCESS in 16m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e524287ac7e64f749aa949cc24b45fb4 : FAILURE in 16m 51s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/684edcb7da7640509906f21be88a6a49 : SUCCESS in 11m 33s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b0e0a0a3340464c99a9255684ff59b1 : SUCCESS in 14m 03s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77f182a0565f40f49211ac3b04e7b248 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f5c34367c1014c2afa3e1d1a92611d8db76cc615","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-05 19:37:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28c5bd6f1da2ff74e023cfe8b626eeb369fa0a1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 20:53:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7ce2372ebe21499182843101bc3bdf9a : SUCCESS in 9m 50s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f166fa6efb924d5387d203d40486b48a : SUCCESS in 11m 03s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7a30617e59b7422aacd15cdd4eeee222 : SUCCESS in 13m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37a4342e37204d5bb21d317735b40b62 : SUCCESS in 11m 53s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89fd3958d2fd42bf82e191feb466b7ed : SUCCESS in 11m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d129672b404a12955336b6585173a2 : SUCCESS in 7m 12s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/09a83f8cd52346bd9f3c9d13a3009a43 : SUCCESS in 20m 53s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b71587407ec74ef5a76e79bac654c1da : SUCCESS in 15m 30s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c0ecee7e89e4461ae4931f882b456be : SUCCESS in 16m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b08795c1d1e14afc86276fcded7ba8c1 : SUCCESS in 18m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11bd560c8ca44d24b94ba1092280d6b9 : SUCCESS in 17m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b490f8c83bf48a3bb2095da0d311dbf : SUCCESS in 16m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db07575053a141fba18a8c54311d73a6 : SUCCESS in 17m 49s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db0b75f8a68a4f9abdae426728b0c0ec : SUCCESS in 22m 59s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d603fc1f21df4a6abe981399cfe62db1 : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"495652f5785054f7d4478776f3c8637c72208add","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-06 06:29:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ca5a3774d5cc058681f5a510a44e2105c7fee36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:29:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3937d1c61a3b3d9372ae9e7f52a4ed0bb3dec07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:45:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5e3940aff785433590e29cf5ebfe3e2f : SUCCESS in 8m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f3c99f749e3c4c56a5f0b2e1a00b50f6 : SUCCESS in 10m 23s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd469bcc44be46cdad9ee884770079c8 : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a7bc460dd8452caf94bdb0b500c74c : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"51816f933d5f8980f50c9318c0d526dcbd5f18fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:45:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5067fc126530524c00d9299363f72088fd5103a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:47:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e56e92478b491fa62ca700afce617a : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c957dfef9f7792998e3b82789933f232dc93794","revisions":{"8c957dfef9f7792998e3b82789933f232dc93794":{"kind":"REWORK","_number":1,"created":"2020-09-24 16:54:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/754155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/55/754155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/754155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/754155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/754155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/754155/1"}}},"commit":{"parents":[{"commit":"a4f917b909fd3fa0fa5d1f40b6681f56cfa76820","subject":"Wait for aria2c to finish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a4f917b909fd3fa0fa5d1f40b6681f56cfa76820"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:54:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:54:00.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I38f31860408ad878ad57aaacae7e7776fec57dc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8c957dfef9f7792998e3b82789933f232dc93794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8c957dfef9f7792998e3b82789933f232dc93794"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
