)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~755701","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ifed6a818868943c19bbdb3620582c2f2aca3b937","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-linters-update","hashtags":[],"change_id":"Ifed6a818868943c19bbdb3620582c2f2aca3b937","subject":"Fix linter errors","status":"MERGED","created":"2020-10-02 06:32:00.000000000","updated":"2020-10-02 10:54:53.000000000","submitted":"2020-10-02 10:52:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755701-1601635952762-c1ef629e","meta_rev_id":"800c9707909809b30a011fe63ea30e02ae51f9d7","_number":755701,"virtual_id_number":755701,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 10:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-02 10:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-10-02 09:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-02 10:31:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 09:51:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-02 10:31:12.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-02 10:52:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e0551956e589804deb00bbd134e1cd78ba9952a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 06:32:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"674884f9ef8c18c23d888016d1ca7204c5213563","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 06:42:53.000000000","message":"Topic set to osa-linters-update","accounts_in_message":[],"_revision_number":1},{"id":"49f3cb9b2975069e0938d4cb80b911b9eb8cf4a6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 06:48:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"149aa570add810d0f09a68615faf6427d9f8bcb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 07:05:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6b3862d83bdc40df96945a900f7f74d4 : FAILURE in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587ea9674d99438c9d63d22648758e01 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f2e30c79018a9e266c613c36bc3558f896dd7cdd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 07:07:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08cf66ef0cc9c54006cdbf19715d1bf31a0f42d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 07:43:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e660787c729f4f829d9ec092a1b8f51c : FAILURE in 10m 19s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fce3898f8ce74d4197b8a428dc648e0c : SUCCESS in 10m 57s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/1c8ce1f020c446b2a0205ea720ad04ec : SUCCESS in 16m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/137c29f718f945078345761b14ffff5b : SUCCESS in 12m 12s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66e7a1c39e6f489dbae428457548d9de : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8908502e0944493da4231cbeb885f678 : SUCCESS in 7m 29s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/8c71bb6a930d4bd88cf44713886c4b2e : SUCCESS in 22m 50s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95bc2b6e08b24c3dac0aa7a5d9afe749 : SUCCESS in 16m 34s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c874e9f6ec0a4374a0fb595c26ea086a : SUCCESS in 15m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a5e9d306aad4efbaa93bd8045380111 : SUCCESS in 16m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bdb7ec403a044cf9a62216d768127ab : SUCCESS in 18m 04s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72a7db30c59641c1b1407efddfd635ad : SUCCESS in 17m 13s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/122c56298ad44ad8ba2f836428e78a16 : SUCCESS in 17m 01s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a5c0b59de2c417ca6f4e5d81ffb13ed : SUCCESS in 18m 07s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bbd5ca30bba439999705c33ab07adf8 : SUCCESS in 19m 31s","accounts_in_message":[],"_revision_number":2},{"id":"26f1e5278de64c6da84f90f9a37c420cf2352f16","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 07:52:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd72401d29bb823e5666cda669f234e067b68e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 08:09:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ca66adaecde44c2085936f981bc28de5 : FAILURE in 8m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4318145d305342f3b8e5edb1108a65c4 : FAILURE in 7m 23s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/bad5d55fa1f04bfd93a22dfb91ab38a6 : FAILURE in 10m 04s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/435d7fe4e09a4e5b88254d4435a0fb26 : FAILURE in 9m 25s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bb5ee2b5c2747a48697489a8541fd2e : FAILURE in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2276ddc735544f288a64cb3c48301b1 : SUCCESS in 4m 34s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/acd87af6f3ef48cc9fc7100632b66564 : FAILURE in 10m 02s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f7d8c2c25fe43478202544c9546f297 : FAILURE in 8m 05s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19eb950331c945a891f0c3901c5e5037 : FAILURE in 7m 46s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56afbfaae5a0483db8790f231e973325 : FAILURE in 7m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab4411179d3b4283914d0e8b58b64da5 : FAILURE in 8m 40s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/984c51c666d04fa5b7642f0cc18d6867 : FAILURE in 8m 49s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9f273ab793744feaaa67d5f1f6f29be : FAILURE in 7m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0257aa04e9d4d21980fd9c130122860 : FAILURE in 7m 33s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9de60c5d3d2b4b1cb79a40c86c61ec57 : FAILURE in 8m 36s","accounts_in_message":[],"_revision_number":3},{"id":"2039c5131c7dd0c9751d9c4f350f062c510e92bd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 09:40:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"67ee30c79e6795b93e2ca45e6c190b02c06a7ff0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 09:51:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"66bb59738c7f860a022ebc34b8391383a0c6f027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 10:07:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8f216f703d044f69a99d365498c97068 : SUCCESS in 8m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/65d482ef157743afabde9ed43503ed68 : SUCCESS in 10m 43s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a090818f833141d3a08c0c2b83538d18 : SUCCESS in 13m 24s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31c47177fefa41fc9f3b07dd14750439 : SUCCESS in 11m 16s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/740d35440fd94e95af993de152ef71ba : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84530c34e0214cbba81b1b9bae6bb6f0 : SUCCESS in 5m 29s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/3d803e0d582a46299ae648647c34ab27 : SUCCESS in 21m 59s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78b017310a104e0ca107b8c43a76b374 : SUCCESS in 16m 22s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e1c623c11124d069b6bcd8530c01b7f : SUCCESS in 16m 30s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab0f90067750443286c6bc6b2756c535 : SUCCESS in 16m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dc22cb07d004db7b883abb51416b4e7 : SUCCESS in 16m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79ad9cebf4624bc6bfade22e0088a13d : SUCCESS in 17m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aac023028d2c4ca2878813f5f9534935 : SUCCESS in 17m 08s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8314e4c8b0084147877f09b8eead4075 : SUCCESS in 17m 18s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/024e54f8b1fb4a9b8c5c4518439a800b : SUCCESS in 16m 34s","accounts_in_message":[],"_revision_number":4},{"id":"dce74ff768b30d445fcf9d8e59010efeb5aeafbd","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-02 10:31:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7034066889b2a22431a61dd98dacfe4184cf4c38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 10:31:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"21a397a5ac57392bfc54903593111669c51bc4e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 10:52:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9d0730b97f3243218b9b1faaf0b3200a : SUCCESS in 7m 58s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/dabaab0d134c44608523f4bd8abe8170 : SUCCESS in 12m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b770a8e4818647f3a3098b8551b7f630 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/535f457db78c4a48bba5d43e2a28d253 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":4},{"id":"26efad510b8918492b13e7a7fb3f0ac886664402","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 10:52:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"800c9707909809b30a011fe63ea30e02ae51f9d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 10:54:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310f5dc5bed64f688b266c413fb7c357 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"01eecae8d3e728f44ac7cbadf23f2f5f2c3c07f3","revisions":{"4a9f6744d96a4ec5dd35f6b559fd17830dc7397b":{"kind":"REWORK","_number":1,"created":"2020-10-02 06:32:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/755701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/755701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/1"}}},"commit":{"parents":[{"commit":"b8e07fca1a591b91f5a16a47a7ecd1bb2688b3c9","subject":"Merge \"Remove support for Centos-7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b8e07fca1a591b91f5a16a47a7ecd1bb2688b3c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:31:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:31:48.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Ifed6a818868943c19bbdb3620582c2f2aca3b937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4a9f6744d96a4ec5dd35f6b559fd17830dc7397b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4a9f6744d96a4ec5dd35f6b559fd17830dc7397b"}]},"branch":"refs/heads/master"},"c33af7b541c70f13cb359cf4feba6fc7ea740448":{"kind":"REWORK","_number":2,"created":"2020-10-02 07:07:27.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/755701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/755701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/2"}}},"commit":{"parents":[{"commit":"b8e07fca1a591b91f5a16a47a7ecd1bb2688b3c9","subject":"Merge \"Remove support for Centos-7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b8e07fca1a591b91f5a16a47a7ecd1bb2688b3c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:31:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 07:07:15.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Ifed6a818868943c19bbdb3620582c2f2aca3b937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c33af7b541c70f13cb359cf4feba6fc7ea740448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c33af7b541c70f13cb359cf4feba6fc7ea740448"}]},"branch":"refs/heads/master"},"5567350f6fdce89a9c992c1ee65c9de5fcd0df0d":{"kind":"REWORK","_number":3,"created":"2020-10-02 07:52:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/755701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/755701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/3"}}},"commit":{"parents":[{"commit":"b8e07fca1a591b91f5a16a47a7ecd1bb2688b3c9","subject":"Merge \"Remove support for Centos-7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b8e07fca1a591b91f5a16a47a7ecd1bb2688b3c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:31:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 07:52:14.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Ifed6a818868943c19bbdb3620582c2f2aca3b937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5567350f6fdce89a9c992c1ee65c9de5fcd0df0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5567350f6fdce89a9c992c1ee65c9de5fcd0df0d"}]},"branch":"refs/heads/master"},"01eecae8d3e728f44ac7cbadf23f2f5f2c3c07f3":{"kind":"REWORK","_number":4,"created":"2020-10-02 09:40:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/755701/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/755701/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/755701/4"}}},"commit":{"parents":[{"commit":"b8e07fca1a591b91f5a16a47a7ecd1bb2688b3c9","subject":"Merge \"Remove support for Centos-7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b8e07fca1a591b91f5a16a47a7ecd1bb2688b3c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:31:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 09:40:16.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Ifed6a818868943c19bbdb3620582c2f2aca3b937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/01eecae8d3e728f44ac7cbadf23f2f5f2c3c07f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/01eecae8d3e728f44ac7cbadf23f2f5f2c3c07f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
