)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~756440","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ice244f946bb2f7199979e0d74ceb61ce64d7a68d","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"Ice244f946bb2f7199979e0d74ceb61ce64d7a68d","subject":"Add infra scenario CI jobs","status":"MERGED","created":"2020-10-07 06:44:29.000000000","updated":"2020-10-13 16:23:45.000000000","submitted":"2020-10-13 16:15:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756440-1602605724780-81340386","meta_rev_id":"5a123d255e454b32fa858daec843e637582309f4","_number":756440,"virtual_id_number":756440,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-13 16:15:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-07 11:07:41.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-08 07:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-10-08 07:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-08 07:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-10-13 13:04:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 07:56:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-13 13:04:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-13 16:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7485bcfaf3ae29466dd74a21ef5495ef22234560","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-07 06:44:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"078721a511044005c07c33d8534e7389357aae3c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-07 07:11:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cec5464a96d99e9c469ad2ff850a1905ef36d036","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-07 07:50:17.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6473ad6ec156349ec8a2427af4d1df1b3081e9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 09:15:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/638f4e199bcf4339b3da3710713285ed : SUCCESS in 10m 45s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/284b0227c98146098db60f373333a99b : SUCCESS in 11m 38s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/30637f57838d4717b004d2a59a877908 : SUCCESS in 13m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b003f4aec47947408847023db823fd2d : SUCCESS in 10m 50s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ed8d006610f4e01b339e07cea801121 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3424824e2ce84e679cd0abfbc705e456 : SUCCESS in 5m 09s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/5246e231fdb4436398bd30811072da80 : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/cbf17d3c593349a59c03b737340b8bca : FAILURE in 46m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee7afd253ef14430ae7a2200a64f68cc : SUCCESS in 1h 08m 48s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/bc13e0343f7a44529f64d10c41eac528 : SUCCESS in 19m 46s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/322b923fe649466cb12a1729bc44a226 : SUCCESS in 16m 35s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40b86187cf2e4b1b905cc89c487d3777 : SUCCESS in 16m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/553f10ce6e9e4f4aad95672010a06b85 : SUCCESS in 15m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8dff026f5ba44c1abcdd5da43103baa : SUCCESS in 17m 07s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/261866a0dbde4155ad4c2f12aa8a7d19 : SUCCESS in 15m 29s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb6826ced4614a2aa4576a2f89fc25a8 : SUCCESS in 15m 58s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fd7f1e564ce45058c5ed0ebba37c182 : SUCCESS in 18m 45s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/189b6346ecd9447180d9ba00bca542df : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":3},{"id":"70a8efab7dfe23b18810347d4b4926df8aa4f55c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-07 11:07:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8d8df5140706c99b07439e234156f578ab5f692f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-07 12:15:06.000000000","message":"Patch Set 3:\n\nnah, that\u0027s reproducable error and I\u0027m trying to look at it","accounts_in_message":[],"_revision_number":3},{"id":"9f05345fd22a6b936e7f7134317fa725e9f08cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 12:42:35.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c0ec2a9b0cce46b0abe96cd27eb8da9f : SUCCESS in 11m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cfa3875f794f4d27b5db1997b0314e2c : SUCCESS in 10m 27s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/694b8951d90f4f9192f7cb28a5187a71 : SUCCESS in 13m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21cd8113a0b34faea228a0f7821ef83b : SUCCESS in 16m 16s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2a87a1de9f041fa874aa54786ed2fd9 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde274f0678d44c0bc017dc143d26c2c : SUCCESS in 5m 20s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a1c404e8ff684de6bd4d9b1aa7e57ea4 : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c6c77cd7ecb84997823202d65ac8bf36 : FAILURE in 56m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efc7c7ead36b446f930f28145a2c90eb : SUCCESS in 1h 00m 29s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/e801b14ee07e4be0818600485a9212cd : SUCCESS in 28m 51s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d18d563a4c3e4b29a1953bc44073c1c1 : SUCCESS in 16m 33s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61e2c83df8a04723a70245b8b86c4aa0 : SUCCESS in 15m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf7c5454634d434ea1602711294d223c : SUCCESS in 18m 30s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2df61448ff240d8ae700dd81212e1ad : SUCCESS in 15m 20s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e23dcebf16a845bba839ba4cf2c514a2 : SUCCESS in 14m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de61b2a6dcd44d2a8734a84999bfec50 : SUCCESS in 18m 38s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a24947bf8d8845628dba8bde8aa76698 : SUCCESS in 18m 29s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86ee7e529c7744efbca66800de4f674f : SUCCESS in 16m 37s","accounts_in_message":[],"_revision_number":3},{"id":"851586bd5a0f89d008406f368f0f6247e721b8c7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-07 14:46:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d1741090a868ad15e32aa36d7280f3b4b6387d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 15:31:25.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c9cd8225622eb17d72c09ac181360899c9f9751f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-07 16:57:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8454a1a57fda08209d18a118ef612462d78f5352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:26:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3bdd9800bfaa46f4943003fcbbde36bc : SUCCESS in 9m 15s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/77eda188b1114592895ae9cbac3ea0e1 : SUCCESS in 10m 22s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/e703df057450401b935f2381ffa628f0 : SUCCESS in 14m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a83f595aaac4e9ea1940e93b97f58a1 : SUCCESS in 10m 21s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d4e04c2c6f34dcf860176a09a8f793f : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29dbd82c3074f3bb70dce7b8f9f9089 : SUCCESS in 6m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ad107e0741264b93bfae7e2ee631b6ff : SUCCESS in 1h 15m 52s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8dde0d94f6004b02a477f47dd05f03cc : SUCCESS in 1h 04m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a42c59189c492baa26fad04bf08cbe : SUCCESS in 1h 00m 51s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/dcdf49a62ccd4fb9995e40660d98f906 : SUCCESS in 20m 27s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b2e6a502c30479fa1fb9bb599a25c57 : SUCCESS in 16m 49s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2ccf8b584974495bd5587ccfb8e70c4 : SUCCESS in 26m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30d29074bc22419eaa51547e00b8403b : SUCCESS in 16m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d66cfd8a54248f4a79ea4c44eb7ab53 : SUCCESS in 15m 20s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b43374600944e9cb00201a071eebc43 : SUCCESS in 15m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82c6ffb78c9f4666af45aa064d893b78 : SUCCESS in 15m 31s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50f760ea617f426798aa194981df2e70 : SUCCESS in 20m 15s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfd8448cc84a4fe180bb1da0757f3f04 : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":3},{"id":"0508945034759ab1f0b3ccfb95c14648fa5a996d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 07:51:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2e03f1765cfb01523807069dc50bb49b5f43a0d3","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-08 07:56:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7fe255eaf636634efdcc953be338a41126647ef3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-13 13:04:46.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4c07dc560d0213af83c72b50610a19783406febf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 13:05:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2f49e2354cc1370efe28d44b6c1e831fec2289e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:15:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"030e676ed5a60ab13f79d64c0f637010f0a8210b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:15:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/04a8ec0ce75c4c5a9b8a85c85e51c92c : SUCCESS in 9m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1376b18f75f14d389f2d3eddf08b3ed6 : SUCCESS in 10m 43s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d868b6231624555b01315a41d4030fd : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7e478c15044aad9685ce9aa890e6a5 : SUCCESS in 5m 20s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/868461df5e1e45d3b7002a3f035e808f : SUCCESS in 1h 11m 30s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/eb39573f076b42ea95d03f6849f0e043 : SUCCESS in 1h 07m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5988b45cc1ef44c790cf5dc8bcd545ec : SUCCESS in 2h 55m 10s","accounts_in_message":[],"_revision_number":3},{"id":"5a123d255e454b32fa858daec843e637582309f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:23:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f27652efbb4dd8ae8b071fcc9441f4 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"70e4d1eb15bdc0a5874f0c5eda0e13a9739c648a","revisions":{"35c3c70dd492aef74e351a428f9596a22c91b2b2":{"kind":"REWORK","_number":1,"created":"2020-10-07 06:44:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/756440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/756440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/1"}}},"commit":{"parents":[{"commit":"7e76855d9f2d21ae7bc3ce22c273939584fb8c59","subject":"Replace some base packages removed when minimising the base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e76855d9f2d21ae7bc3ce22c273939584fb8c59"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 06:42:38.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 06:42:38.000000000","tz":60},"subject":"Add infra scenario CI jobs","message":"Add infra scenario CI jobs\n\nThese jobs will test the role in the context of the integrated repo\nwhich may show errors not covered by the functional tests.\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Ice244f946bb2f7199979e0d74ceb61ce64d7a68d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/35c3c70dd492aef74e351a428f9596a22c91b2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/35c3c70dd492aef74e351a428f9596a22c91b2b2"}]},"branch":"refs/heads/master"},"c460762bfdc801b7e59987ddfed1081f5dbd0daf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-07 07:11:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/756440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/756440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/2"}}},"commit":{"parents":[{"commit":"93df10df616ac653a3a52c1eca29630bfcdaeca8","subject":"Replace some base packages removed when minimising the base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/93df10df616ac653a3a52c1eca29630bfcdaeca8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 06:42:38.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 07:11:32.000000000","tz":60},"subject":"Add infra scenario CI jobs","message":"Add infra scenario CI jobs\n\nThese jobs will test the role in the context of the integrated repo\nwhich may show errors not covered by the functional tests.\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Ice244f946bb2f7199979e0d74ceb61ce64d7a68d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c460762bfdc801b7e59987ddfed1081f5dbd0daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c460762bfdc801b7e59987ddfed1081f5dbd0daf"}]},"branch":"refs/heads/master"},"70e4d1eb15bdc0a5874f0c5eda0e13a9739c648a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-07 07:50:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/756440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/756440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/756440/3"}}},"commit":{"parents":[{"commit":"93df10df616ac653a3a52c1eca29630bfcdaeca8","subject":"Replace some base packages removed when minimising the base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/93df10df616ac653a3a52c1eca29630bfcdaeca8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 06:42:38.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 07:50:15.000000000","tz":0},"subject":"Add infra scenario CI jobs","message":"Add infra scenario CI jobs\n\nThese jobs will test the role in the context of the integrated repo\nwhich may show errors not covered by the functional tests.\n\nDepends-On: https://review.opendev.org/756244\nDepends-On: https://review.opendev.org/755497\nChange-Id: Ice244f946bb2f7199979e0d74ceb61ce64d7a68d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/70e4d1eb15bdc0a5874f0c5eda0e13a9739c648a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/70e4d1eb15bdc0a5874f0c5eda0e13a9739c648a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
